)]}'
{"id":"openstack%2Frequirements~949883","triplet_id":"openstack%2Frequirements~master~I351f8265e0277140f6f3074105223173d93d3042","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I351f8265e0277140f6f3074105223173d93d3042","subject":"update constraint for glance_store to new release 4.10.0","status":"MERGED","created":"2025-05-15 12:14:58.000000000","updated":"2025-05-15 18:45:27.000000000","submitted":"2025-05-15 18:44:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949883-new-release","meta_rev_id":"6c76226a610f7f41e5030d6e98e0007a7fb105ad","_number":949883,"virtual_id_number":949883,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 18:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 14:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-15 14:42:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 13:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-15 14:42:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c7f0b4dd1827dff16f09eedcc9c791c32900a07b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-15 12:14:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c87206dd9414f823f8fd1fd0f53871c09e58276c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 13:07:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45ed4d6de126463ea5fbbc8387c3399a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de91763c32141ff8e819d35a3f370ec : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22408f3d99034c64b669e30b8f96daf2 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2603c9e776fa42c8a29c0d7a26d50d96 : SUCCESS in 2m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93cdd8866cf04f7782520e6f5b18dcca : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b741ec723ee48368be4b513fd55ebb6 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/625e10bede9c406791f36f94c4b28c25 : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af52cf061f7144eda2cc45be7bf61a3a : SUCCESS in 3m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a5b365e5fad14d3d8c9404abddd96543 : SUCCESS in 15m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/443c8e0eb2094e869c0e4be44935159f : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/41b88295eab845b78cfbd25239952dc1 : SUCCESS in 9m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dc701e4fcccd4a6ba8a73d05db200943 : SUCCESS in 10m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/be28d1b0753949d5b28a625765174a69 : SUCCESS in 12m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/690757e0a7694e2298ffae99a2baf6f4 : SUCCESS in 4m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/747f825791394c338092463159751d4c : SUCCESS in 5m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f44013485daa4c86ac1e0bbb5277fde7 : SUCCESS in 5m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/29b223b2ef9d4e699d5241d893817799 : SUCCESS in 10m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6d5358a52aea4350a205f1172696d815 : SUCCESS in 5m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9fcc791b6b264edb8b58d21fc1866093 : SUCCESS in 6m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5b6e1ca75de34f80b9087504bcf4dc4c : SUCCESS in 8m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ba51763a84534ab69656aa347bf1b389 : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ac69e3f81e24785a5d389a1b4136375 : SUCCESS in 6m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/883be8d370e0401a86930379460095fe : SUCCESS in 9m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1cabe06c2fe349678eaa13ae6ac22f5d : SUCCESS in 11m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4167099ec93b4602b78972be30fb19d9 : SUCCESS in 3m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bef7b25946bd430d927f33bab2c092c2 : SUCCESS in 9m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e30c8d005f4548b69568fec64e47f8f0 : SUCCESS in 6m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/56f62c62ad2f413893bc7b4ce301f725 : SUCCESS in 2m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/90c9f022ff964270b3f46837ba0f72fe : SUCCESS in 9m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ab2804f23c8d4d5c980bec6419e51954 : SUCCESS in 18m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6848d4c1d2db4d3797f534320a258cfb : SUCCESS in 6m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2b4939808c5841e4a7b3a2262b7152ad : SUCCESS in 13m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b8e45f10444744de9db9afbd827d97d7 : SUCCESS in 2m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fdd803e74a914474b05ca7602a3f9306 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/529dc6b67b754277b820305fbf33b85d : SUCCESS in 24m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6f47e65242f54b1c94be0c80076a9c24 : SUCCESS in 3m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cf4206c511c7423694e5fab5e1f103e6 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74184733016145baad9fc7318dd2751b : SUCCESS in 6m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b3fb63e0cf3142fc8cf12bf4023827c5 : SUCCESS in 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cc98277077b4fdcb99e1e09542ab189 : SUCCESS in 51m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f71a3dbbb18543d782cb0b3e42005c7b : FAILURE in 47m 32s","accounts_in_message":[],"_revision_number":1},{"id":"94429aa8aa10fcf2ca4af996b237fa41c1b74185","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 14:42:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d08a0d62af0bf1871376a24a2a0fe8c3841f428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 16:39:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d0f22247b4f4512a3682ba1722feb2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853ec716946f4779b04412f24c0600ab : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98cc25b6d84f42909ae0608e44b978da : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc455c9ef34b4b89ab0c6e3f4ca8cc43 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e2f18a4b66d41be89e936dc8e62b62a : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f72d11c20aa449efa18ca5ee74a31af1 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4a9db79cc864fbe8d9cfc6e60e8f78f : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3e9fc05d58e41319a713dc36c1fa7c0 : SUCCESS in 3m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0369f41218d6424ab117b9908bab22e0 : SUCCESS in 6m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e76061d11bcf45929de4fdefcee31931 : SUCCESS in 10m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/42e6c6f4db7a4d8fb9d4d3ea143723ce : SUCCESS in 15m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9930f1d0daf04d9f85b9fbe27444ea08 : SUCCESS in 19m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/528d347a48284df0bd3554a3cd3afa5d : SUCCESS in 12m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d09e399aae0546818845a04a46c27945 : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5a0994e7639e4cd0b96e83138209732a : SUCCESS in 3m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9c80d49e037042478ff1cf588f28947c : SUCCESS in 4m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b4155283e30248e3add64f086c8b1ff5 : SUCCESS in 7m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4b782672f97848378a8b190bbb593d9f : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fdc575c472984e03ba05c900d3829d7f : SUCCESS in 3m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d7d2c68ac1fa473e982def57dfa7a9df : SUCCESS in 7m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e084a9028b0c42f0badc98e4a5eafb13 : SUCCESS in 5m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18036aa029cc43439fcc709e995b8337 : SUCCESS in 6m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/577a789b64b34f8aa22711c922dc7d08 : SUCCESS in 7m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ce05b26e25b94d3a909474ff3808b137 : SUCCESS in 9m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/45c45cf788c44c6995e39e1ec649b7ac : SUCCESS in 1m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/07dff097b43040ae900a0dfa5415d0d5 : SUCCESS in 9m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6c6669323320455cb26bef930b2fc517 : SUCCESS in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/779b5f9c25674d06a3ca827c4077cb8b : SUCCESS in 3m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/32e4a7e2a8aa4ee2a972d33b6b378d8c : SUCCESS in 10m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b24c6b02ae7548d6ad4a0ba7657294fb : SUCCESS in 14m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ecaba0798f3f40ae975b443aea37b6cd : SUCCESS in 7m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/09e468505509417ba2b888d1fd82304f : SUCCESS in 10m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/54d151e61dbe47cc8f7d1a994eecd9d3 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/914f02bc60fb4f8a88872726eb85e5d2 : SUCCESS in 2m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/452fbc21249748cfba578b1a9097fb34 : SUCCESS in 28m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8e5ce06027b14dd2a9399f74abe2b55d : SUCCESS in 4m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a97c1d2d92304c178d3983cbc2000d2d : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a0ac1247b83494b9a4720bf0d5d4e4c : SUCCESS in 3m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c28b28de64b84ab3abe0dc762768e681 : SUCCESS in 17m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cf08b4ea80043538fa6ccac3e8a18cc : SUCCESS in 38m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5117027d3a994971a51ca2a65273f5a2 : SUCCESS in 1h 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e5313e40a4b22f809216367439b129b7c0878961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 16:40:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa1cad4433f422e6da20af9ab5cb8cd07afe709c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:44:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d92ed334a7429993a45fc44ce3afb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd55401ddbf4364bd48807b24c348e0 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23863c8438aa4566aeaf267f16a3798e : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ec12c732cb14f14a6457e64b1c2f949 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68e1424c923e4fc5ae1fc93fcbcf4dc8 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a56e7c566c945a69dad364d3eeced3f : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f74cffd6a464aadbe8e94f2e0b329dc : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a03239027fd042a6b557e88d920fb003 : SUCCESS in 17m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/031d6967d874407abe069e87f2d18395 : SUCCESS in 4m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/13369a91c5aa4b678ab6c86dbf0a8e61 : SUCCESS in 18m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1ac60b8f5fea449680fdb104191d90ed : SUCCESS in 9m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/171b7bf2be064e8094c51a6dc0bf3cc2 : SUCCESS in 13m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/37b0a3661d8f4f939c2dd63e6bc1c630 : SUCCESS in 6m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1ec7a7be260f43b1b6479a92ed1c91d9 : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/091883240ce04cd899b0825d10bf53ec : SUCCESS in 9m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/93784a2c49aa4debab3497707c82d0a2 : SUCCESS in 13m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/622b4fa60aa44c08997e810db13b694a : SUCCESS in 7m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4fb846b9424d4be39faa81772c4e09e3 : SUCCESS in 5m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ff640b6f7ab14a0f86f01d87fe909512 : SUCCESS in 12m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/be7cd1df047745c1adb849892bc54e09 : SUCCESS in 3m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/da3de8a8782a436baf188ff65cfb1266 : SUCCESS in 11m 28s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8c0d1333534d4aabac0f9e1c75226e0a : SUCCESS in 19m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2a5a92e7e80a47998458c673a5fcc5f4 : SUCCESS in 8m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/310c548a66934f27b3a1c3a333949da0 : SUCCESS in 14m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3d0ea74385374a2ca46fa9552242eb04 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c78a93560bd642a3a498c7d0d88be21e : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a30efbf8284143acbeb1052a3e1e58d7 : SUCCESS in 23m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e5c021014288412db37cafb01be439ec : SUCCESS in 6m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a97a1e3a1b4f4613b3c141b5e065c4c6 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/741f52a0f95f43ffb0ce3f700d01ed44 : SUCCESS in 5m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e750a90f637042f785c0f015d8cb942c : SUCCESS in 7m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70886855966c41b2805eb4bb7b6866aa : SUCCESS in 55m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36379015e8064f2d80efd1e2740b4f88 : SUCCESS in 1h 51m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9761b683358bcc23b80a9318cc839379837afe4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:44:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c76226a610f7f41e5030d6e98e0007a7fb105ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:45:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f54f1c65e34471a2802fe4bf4032a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b9144d433c4701affd8e634ee48c34 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef59f25e24ca542c234a0414443eadf805bffb55","revisions":{"ef59f25e24ca542c234a0414443eadf805bffb55":{"kind":"REWORK","_number":1,"created":"2025-05-15 12:14:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/949883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/949883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/949883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/949883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/949883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/949883/1"}}},"commit":{"parents":[{"commit":"13fb430bbda5f25fc7cf294d1511c66ec032bb2a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13fb430bbda5f25fc7cf294d1511c66ec032bb2a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 12:04:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 12:04:56.000000000","tz":0},"subject":"update constraint for glance_store to new release 4.10.0","message":"update constraint for glance_store to new release 4.10.0\n\nmeta: version: 4.10.0\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Cyril Roelandt \u003ccyril@redhat.com\u003e\nmeta: release:Commit: Cyril Roelandt \u003ccyril@redhat.com\u003e\nmeta: release:Change-Id: Ib1ce5802dc996ecf899dd8558a14b962b54d5393\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I351f8265e0277140f6f3074105223173d93d3042\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef59f25e24ca542c234a0414443eadf805bffb55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef59f25e24ca542c234a0414443eadf805bffb55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
