)]}'
{"id":"openstack%2Frequirements~949889","triplet_id":"openstack%2Frequirements~master~Ib1b525105f1603c29621ffbe756bd7bacc64c7d2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib1b525105f1603c29621ffbe756bd7bacc64c7d2","subject":"update constraint for sushy to new release 5.6.0","status":"MERGED","created":"2025-05-15 13:20:10.000000000","updated":"2025-05-15 17:11:30.000000000","submitted":"2025-05-15 17:11:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949889-new-release","meta_rev_id":"1af11474f146f9ebac2ba6f0adcfa2cefceba35b","_number":949889,"virtual_id_number":949889,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 17:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 14:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-15 14:44:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 14:44:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-15 15:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a28173d2dc5c02467cfc6e0d3b5954e6b48dea4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-15 13:20:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19cd3534f088fdc6ad95c6e68618e7569bf096b9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 14:44:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c588aae341ae7fd8be4f2b7273a0019d5c1f2572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 15:05:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fb6d65b382c49c08f7c70dc67363f11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87822f71ac144a27a5eef0a2d08fda23 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b6d99987984b43bcbbdf89e5af78d7 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddf5f26b7e4e460fa6d25121f64910b5 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9332f95984df493186197e51a0801276 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1af17342397a4d2cb57cd9986d524372 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/825e4c68393e4c30b97c2aa357fa6793 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2f8ff2240fc4a4f9426afb9c7eedee8 : SUCCESS in 3m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/65c042ac21474158a25b70195c2292d5 : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/49c90b4ac6cf414e98060b508e879847 : SUCCESS in 10m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/131364e517b146c49fbf75b3aa0482bc : SUCCESS in 17m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/103a4a54f69d4f9ca5ccedc643e05128 : SUCCESS in 20m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b833ec0d4ec04283be37d5e18c089047 : SUCCESS in 26m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f8740580d576499190a2446fbd58ab06 : SUCCESS in 4m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/20ced25969884f24b0be3a1ffc33de3d : SUCCESS in 5m 02s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/041dfa8e944549e1b27d1f99ff57d477 : SUCCESS in 3m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8641a7b1dead4391a91eda44ff16aa26 : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/97f2a671d0434bcb90ebceff6f46796d : SUCCESS in 3m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3f5aa7957e9340b184d51057c3e539a6 : SUCCESS in 4m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aa5c6229004d404aa2a07bffc186bbe2 : SUCCESS in 6m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ef64bd7652bf46a8baf6f95a2eaa346e : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8bd1f814fac42ad91c042565c7e2737 : SUCCESS in 6m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a6742c725993493987bec1fb5689054d : SUCCESS in 9m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0fb185f9dcd3496fbaa449c0c8c156f8 : SUCCESS in 7m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d6b9e2922de74a9eaa4a6e18819346f0 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ba976e553f5f4ecbb3f08bb900b3b98a : SUCCESS in 7m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/68d030e1192d44d69f0d86970ebd58ca : SUCCESS in 4m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cebe02c4694a4919934b05f79121d711 : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1c33e1e371a44e3db79bd6ebfeedeed7 : SUCCESS in 10m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/655b2de97cdf4e50b7ad67b0b5d90d11 : SUCCESS in 20m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/76226dd1a8114e70887c541e70e211ae : SUCCESS in 8m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6248734e896a46279b7ff6526e6ef331 : SUCCESS in 10m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a8c7479146d9494990d7f9b4fa670f50 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9b81f81df9341269ef102a5ead1b6e9 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e77c193c943548a5be8b66fec4137a3d : SUCCESS in 24m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d691ca44445946538294a39f431626e6 : SUCCESS in 6m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cb943d4e1b8e46e3a36d7d4290a4b4c1 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e1ba7516b484d45a24770b6843ebc04 : SUCCESS in 3m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cc6f61eb9232470788ca3690aa4a0f5b : SUCCESS in 17m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9906a4f0c13a4cbdaa973b734e30d3bd : SUCCESS in 55m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3199d472564a40a8b0d7f5eabc94636e : SUCCESS in 1h 40m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b082f628dbdf8e4f17237218110a3f6769699e8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 15:06:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3464dce0529ac3d380a940bb4e22e479cb2be5dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 17:11:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1df64f59a2b54013897768b238de7407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c12efdc193e4274916b611ebc19b201 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b25540b065146dbbafa7e9905570ac0 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e2da29f3c584e8f96151a3b329813c4 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8546503d5f944a19388492745412e0f : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a287b25ae4304b9b82ddf8652fe77c9a : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f830174c1710443999f49cd5b96779b0 : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d146aec2a43a484f95fe31fc6aa93a15 : SUCCESS in 14m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4badd9ea636c40679ec07b4cbcb453f6 : SUCCESS in 11m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4695544b11d7439690c247001bb8952b : SUCCESS in 21m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0bce42d41d7141029ac96e4c460e5b1d : SUCCESS in 25m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/26ce3db60a3543e184422fbfe539f80f : SUCCESS in 7m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ca27f5eb045f486d9f89d0c6f7723f59 : SUCCESS in 6m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f994bef0140c4b66abf3bfe7843e1205 : SUCCESS in 7m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7106ab6984bc4d43a1351da7ca6329b5 : SUCCESS in 8m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d3b99a733642426faa3f63b87661a88e : SUCCESS in 5m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f475241afd74f1f9d4b00457b84c6da : SUCCESS in 3m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/32242e69306d4f53b300052c890607c9 : SUCCESS in 4m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9dfcd6f548484c178fa42efceef85916 : SUCCESS in 13m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/03e663d100d64ec8b127a11d47521a93 : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8b87b3bbae4f42cc8966d8ec73fb9c71 : SUCCESS in 7m 55s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/82c8bc09998a4e54bcc14e94d4c5b4b6 : SUCCESS in 20m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7600ea4d24a6447c90a9912668d9ee0f : SUCCESS in 7m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/09edcbe2b1e343c7823b6bcad11997bd : SUCCESS in 17m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f855b7b5f4e841aeb5668ca923086303 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57ca76a289404aa9a7cdd8f113aa2c33 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca071d2231e84bf48ea8283fc4ce418b : SUCCESS in 24m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/066a5444834648e9ad279c585cd38d83 : SUCCESS in 4m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ead421e9c5474d089d6405882e8066fb : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53f4f81f0f7149719562f32770ab4511 : SUCCESS in 5m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b40f61ecc31d432faee5f34afd89cb83 : SUCCESS in 16m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9a1e5ac2aa543ae97e9194dd9014b3a : SUCCESS in 59m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95186175778f4e179ee09bbfa33467ae : SUCCESS in 1h 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1af11474f146f9ebac2ba6f0adcfa2cefceba35b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 17:11:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6b08462cee919d0dc5d1073a438d90dac0b49a7","revisions":{"f6b08462cee919d0dc5d1073a438d90dac0b49a7":{"kind":"REWORK","_number":1,"created":"2025-05-15 13:20:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/949889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/949889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/949889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/949889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/949889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/949889/1"}}},"commit":{"parents":[{"commit":"13fb430bbda5f25fc7cf294d1511c66ec032bb2a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13fb430bbda5f25fc7cf294d1511c66ec032bb2a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 13:10:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 13:10:09.000000000","tz":0},"subject":"update constraint for sushy to new release 5.6.0","message":"update constraint for sushy to new release 5.6.0\n\nmeta: version: 5.6.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3f9a82038e4c25b2c87d5be163fd9b72a0063581\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib1b525105f1603c29621ffbe756bd7bacc64c7d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6b08462cee919d0dc5d1073a438d90dac0b49a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6b08462cee919d0dc5d1073a438d90dac0b49a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
