)]}'
{"id":"openstack%2Frequirements~949895","triplet_id":"openstack%2Frequirements~master~I8a60034c1b9c08822d687e3d680c9e91cdc58b9c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8a60034c1b9c08822d687e3d680c9e91cdc58b9c","subject":"update constraint for ceilometermiddleware to new release 3.7.0","status":"MERGED","created":"2025-05-15 13:52:38.000000000","updated":"2025-05-15 17:13:12.000000000","submitted":"2025-05-15 17:11:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949895-new-release","meta_rev_id":"efc1997f1f81efcb151f4be43ec839c819a93f10","_number":949895,"virtual_id_number":949895,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 17:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 14:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-15 14:44:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 14:44:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-15 15:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2a1d599d32752f93842697b5c735d810cab4a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-15 13:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8ed5dbad3e7f862da7f3da6d2644d5c1e099a15","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 14:44:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"349c9cd701e7c49e022f2ee8e5f10ad4666dcfc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 15:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0487554a39aa453f836fa98d77292e81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84411954b5064f3381c4cd219bfef6bc : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d630b34e4884f4087f51a5ff9df4a80 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e86668e6d77f4d2d8fdb3a111f8358b2 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a447697725ae43c3a6fe87fe55dbed49 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3df8c0667ba8480d9e3b4767dbbe4b29 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a2a884c343a4e7b8520b804d8b2b306 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae96904b0ed146599d0de064242d4b92 : SUCCESS in 3m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b448c96e4c44c5d9d4197717a105cf8 : SUCCESS in 15m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/24333d64a9444bce966503ed53f74ca7 : SUCCESS in 9m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4a2fd708cebb4ecaa486cce1506549ea : SUCCESS in 18m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/165d8b42fae64448bcba7ae477f9c694 : SUCCESS in 22m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/71465e90d3e84830a17c7e00e751b47f : SUCCESS in 22m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/84dc6f35a08549cc8e13c04eb74681a8 : SUCCESS in 4m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2ec96221890d40b7a6ca4f740700089f : SUCCESS in 5m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1ffe1174fcd3430dbe33bb6fadcbb9ed : SUCCESS in 3m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/06f70531f4b64218a7cef2a627156831 : SUCCESS in 10m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6668f1cb5eb84ec086bfc8f7d72f0c97 : SUCCESS in 5m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/527f31e115194e9b96d13d66f2933dbf : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ff520c324f8643a5b4792a748ab33402 : SUCCESS in 7m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cc2343aeeaa9457e8098b4f57241a24a : SUCCESS in 15m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78cf832ae1124789ba99124dcd8c82c2 : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1baec2752e854fe097fc644aba5779c1 : SUCCESS in 7m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4fd8705702964597ba3f7d3c0a1950c0 : SUCCESS in 12m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/74e697ce66284d07848d90a9ab4bf166 : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d66d057361514143b85d94d9f7a5f473 : SUCCESS in 10m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5e1504e1cb0e477dbac156a4f3c7d54e : SUCCESS in 5m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f0c46af38c0e4d08a280c04a72bc81f7 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/37473d31ef0942c28e979326037dbf57 : SUCCESS in 12m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7a1b2f88fc9143718bba6030e356e760 : SUCCESS in 21m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b1bf5225ced6487ab9d4c4c44c513ea9 : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/75ef7e08ba624668a1cf2ba17fa34dc3 : SUCCESS in 12m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9573ea372ed94d81bbccb1b519558dd6 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52e26f7a9c4b4641ba158e9a5ebdb4d8 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b28d9cfeb6d0455bb12b67a01e930d11 : SUCCESS in 25m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ee1c261862c24b189e546efc24a81945 : SUCCESS in 6m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ab27e0e65c1f4a55a793a6464be25c53 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19b3cde3a6cb4e3c9052da7cb5cb8fe7 : SUCCESS in 5m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dc466567f8ff4dd0a8c7b25500a4fa41 : SUCCESS in 17m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a46adaeefbde472584455586fa7547d9 : SUCCESS in 56m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fbf07ddedb344a9813195802900aa30 : SUCCESS in 1h 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ec0ae9cc9c60e1693c383a52b8a023f190d249d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 15:39:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2eae87bcad604090c149fbd5f65fb9f95223eeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 17:11:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/946c23d48b9b45e297c5d28b2620aac5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41af2bcf23334c9d9decab271d2e8338 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84bdb0f4e12a4a10b097e740497feaa3 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fba0172d06254ef0bd1c4d71ee6177b9 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6223855db8b04fd4b4ce848f5ed66f0c : SUCCESS in 2m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd2c9052461d4deeac998359a46a9dad : SUCCESS in 1m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2a7cd6588234402b77305521561e333 : SUCCESS in 8m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/754c4faaf2714d6bb7607fef3ae5e58e : SUCCESS in 17m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b40bba1878ed4589a1e8edcb266d9bce : SUCCESS in 5m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66f4e077d4e749ffaf5870688cd8620d : SUCCESS in 19m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/df8cbf9f15aa4e81b8a0a9ea32a5d7cf : SUCCESS in 24m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5af1eac26d2a45799768060540fd1e79 : SUCCESS in 13m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0421228e5b504b969b90a71d11fcca4a : SUCCESS in 6m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d55737283b4a4d95ad2f59e3f43ff9db : SUCCESS in 6m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a060983d78f348c698b6237f5792bedb : SUCCESS in 7m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6163e11b188b4f48873e1c2c60b70625 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e2455c20f5a34087a8af25df7c0d658d : SUCCESS in 7m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/05d2b628cc094195b5dbb7367c7cddef : SUCCESS in 8m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9a76348803944752929799e4a0e25776 : SUCCESS in 11m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd8c3b2d99e147ef883139221c0e8f98 : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6902843c542d46d3949f279b0891b5b4 : SUCCESS in 11m 08s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7e3b6a0b82474cb6ad487e8f4daa5cdd : SUCCESS in 20m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0d3a0a7276e54bafb505885e6ec3f369 : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/18fe9570f52f4680898c8dae253c569e : SUCCESS in 12m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7dac6da5d8a1498e86e956606958b015 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/61e3197d507644e5b785b5896430a110 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50a0ad510a854bb1ab7512a5e063911b : SUCCESS in 24m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/148744083eaa4037b91ab4144840f3bb : SUCCESS in 7m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6e2b37d6d0e0417f8268140aa7ca3104 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27333cda6e314425ae6b8ee2b7a2560c : SUCCESS in 6m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/29885a2c2f5f40aaa744cb0afc20527e : SUCCESS in 17m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25cbef1bee1c4a40ad34d3d528dc196d : SUCCESS in 55m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b7bd475349a417ca592cc6dd9053226 : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b6c294adbfefd0113a651d7148469b118ce897e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 17:11:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"efc1997f1f81efcb151f4be43ec839c819a93f10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 17:13:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6923f4c406b4853a84576d2c2fb6902\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94be20f42d3b4bb5ad70ff15762a8c7a : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb40a9f352ea15c9836d53fa406b05cd951fa1ae","revisions":{"eb40a9f352ea15c9836d53fa406b05cd951fa1ae":{"kind":"REWORK","_number":1,"created":"2025-05-15 13:52:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/949895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/949895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/949895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/949895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/949895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/949895/1"}}},"commit":{"parents":[{"commit":"13fb430bbda5f25fc7cf294d1511c66ec032bb2a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13fb430bbda5f25fc7cf294d1511c66ec032bb2a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 13:42:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-15 13:42:37.000000000","tz":0},"subject":"update constraint for ceilometermiddleware to new release 3.7.0","message":"update constraint for ceilometermiddleware to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id400c9be4f647cea3d8f0fc975f6acb0e9240f98\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Juan Larriba \u003cjlarriba@redhat.com\u003e\nChange-Id: I8a60034c1b9c08822d687e3d680c9e91cdc58b9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb40a9f352ea15c9836d53fa406b05cd951fa1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb40a9f352ea15c9836d53fa406b05cd951fa1ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
