)]}'
{"id":"openstack%2Frequirements~950159","triplet_id":"openstack%2Frequirements~master~I6d41cdd860cd45b301ddf3d0246dc6ca39a7c4a4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6d41cdd860cd45b301ddf3d0246dc6ca39a7c4a4","subject":"update constraint for ovsdbapp to new release 2.12.0","status":"MERGED","created":"2025-05-16 13:11:37.000000000","updated":"2025-05-17 22:44:41.000000000","submitted":"2025-05-17 22:44:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950159-new-release","meta_rev_id":"a3bc46f56bdcedf52178cf296944546871fe1e80","_number":950159,"virtual_id_number":950159,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:54:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:15:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:54:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5aaa8b69283f53ec5f4cad6ec312f86e1cbbeff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:11:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fc4cb3751b0fa7e7496d6ca1942152ade442651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:15:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac19176df7ed4ea9981951fc242c9818\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b21537833c47778822578c947c92cc : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8117bb38b9fa4d319765bc8cff431035 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a8f0ef801e14e82899c94e42c928459 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b030da2be9ff4b61b01f6ae7082b8680 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc25255483694b5b9d52d75044580de0 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/929215f17b104b5c9762c5c42963d155 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad9809cc6dbf4b058bc1f428649c9495 : SUCCESS in 1m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e09a10e20eeb48d8908273475b2e335e : SUCCESS in 14m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6ec19d0203f743309a43f9201e7dbc5a : SUCCESS in 8m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fdacd89b6beb472a8de1259927f65e05 : SUCCESS in 8m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e0f8912e4ceb462abb27d7945c02cacd : SUCCESS in 19m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5e13323dafcf45508fce80c248abd1c1 : SUCCESS in 22m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c9e25bae8ca84ad9b693b2cac16ecb3a : SUCCESS in 4m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0cf1a64e2d7948ab812f5eb4151b3d11 : SUCCESS in 5m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/28c8888472d24d57a952c05a5f423444 : SUCCESS in 4m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c3a905d8a065493ea2631e49da5bf955 : SUCCESS in 10m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/911107673d624f3482bf7f851e664caa : SUCCESS in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d71289696eb84b44b64a31d12d3793c5 : SUCCESS in 6m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1ebb60e14a6f48839c79c9dd276b4159 : SUCCESS in 6m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6a82a7a63a4145b9835a0f8b6151bae3 : SUCCESS in 13m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1bc9794dc984e0498213f2058e84a61 : SUCCESS in 7m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4afa046ee1ca46b59b6eb7a634d21fc8 : SUCCESS in 7m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e8fe7b5c32354539b6dd8bbad3229e9c : SUCCESS in 11m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3c50fd6bea7b4f4cbe8fa5a877fef727 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/037defd6753640789234bdf19fc8a5cc : SUCCESS in 9m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/695a5f81c9324a5497e40a2d440bdce7 : SUCCESS in 4m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e0bfed6cd74948f1bc461f4367f35247 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/02537477771c449e9704c00acef89947 : SUCCESS in 8m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c527f86b06c142029c97534f18295fa6 : SUCCESS in 24m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/128580cc368c402e9b3623757762791a : SUCCESS in 8m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9eabc1d226c14a10bc73e1d164de1112 : SUCCESS in 10m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/59030e55ee13463f9046bada61f964f0 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae715d96f662420ba302ad599b199a36 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/073ea3d24fed4713bf20f80c6b90713f : SUCCESS in 24m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6d0cb902abd74f7e894f4a43c166c5f6 : SUCCESS in 6m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/72599f877ad444e081a0e517b882ba6c : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db46019b1e8f4a3dabc8b816be97b979 : SUCCESS in 5m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a5f4fddc88d044f6b04c474bdb62016d : SUCCESS in 15m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2f918b7e5e7406f919c2bef3611a41a : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5be2b91d5d094e82a15e818a3a407762 : SUCCESS in 1h 51m 30s","accounts_in_message":[],"_revision_number":1},{"id":"05c5d80ec05581916ff46ec0dcad93c3729f1d31","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:54:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8578ab067e57886fd66216960e21d88c3912d717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:55:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f33067e48f21695191eb5412ace39ae6dfc900ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360c085d7f2b4d6c9c429cd432baab9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc6a12033e2431a8bf48eb595d197d2 : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a6beb0e36e4509a09ff3200247a171 : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57751629fb47472592a0c4c97235281d : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38c43e24e6c34a9dad5d9c80e168ca5e : SUCCESS in 2m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9134e9294f3942aa95ac5d1e7c75e37f : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee52dc671b8e45d3845b551b94680c6f : SUCCESS in 3m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/560b319223ea4fe0b92d4e3075966360 : SUCCESS in 10m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6055cc76229c4682a5f04ca7aab959dd : SUCCESS in 9m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b14ec40015694269b710373952bfc34d : SUCCESS in 10m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3bcf5242e8414a8aa7828ce1c328ba88 : SUCCESS in 18m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cf2abfb4536347f788a767ccc48d3619 : SUCCESS in 11m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ea734241fe234453803d3fa4a7f36776 : SUCCESS in 6m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e08aae80bf01499a86e1a56cfd171d44 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8daec27d53f547e3a7950b9b331d2fc9 : SUCCESS in 7m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f7cf029f42844b6a841c3e24cd14c789 : SUCCESS in 14m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c760446016694732bd60ca5c980cd798 : SUCCESS in 7m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8a09b8ca78f1442e9cd2811082010333 : SUCCESS in 8m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/758d7be17aaa4660903a3558518f3119 : SUCCESS in 13m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5eb350c046e74cb6b875f02e19ea9a53 : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a2dc4c75ec224815bb2bbd31b75a5daa : SUCCESS in 12m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7ffb9f0280914976b66a081efe3f57fe : SUCCESS in 20m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/786ea2725c34470394afa26e69fb4f76 : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ad64e383652e4317abd555c045a9a8a8 : SUCCESS in 12m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/08131e29c47945efb159f12fbdd5b197 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/662f12c509de4cd49c202b13b016b851 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3ff71fdb79e402caae02c2958ea9002 : SUCCESS in 24m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c27dc1f2fdd34adf89a930191db5b5ba : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e04f8d2611a2473694d30d3bbc3bbac1 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84e446f556774005a5ada3922e223273 : SUCCESS in 6m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/abf2d3972a514d89b992f4e95aa321c0 : SUCCESS in 6m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6975c6c28e3469ba03e987067e38d8e : SUCCESS in 32m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27ee2298e0c944fb9aec6cdaec4eedf9 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a3bc46f56bdcedf52178cf296944546871fe1e80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf8c5e40fc0de643536f1c0a42127f908fb0fc06","revisions":{"bf8c5e40fc0de643536f1c0a42127f908fb0fc06":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:11:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/950159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/950159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/950159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/950159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/950159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/950159/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:01:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:01:35.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.12.0","message":"update constraint for ovsdbapp to new release 2.12.0\n\nmeta: version: 2.12.0\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I6ca2d15385aedd11ae8ff695592c5563f5e65c38\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6d41cdd860cd45b301ddf3d0246dc6ca39a7c4a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf8c5e40fc0de643536f1c0a42127f908fb0fc06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf8c5e40fc0de643536f1c0a42127f908fb0fc06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
