)]}'
{"id":"openstack%2Frequirements~950160","triplet_id":"openstack%2Frequirements~master~Ic0f286bc27f6736406d54753625d9ea2e485f3ee","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic0f286bc27f6736406d54753625d9ea2e485f3ee","subject":"update constraint for keystonemiddleware to new release 10.10.0","status":"MERGED","created":"2025-05-16 13:12:03.000000000","updated":"2025-05-17 22:44:37.000000000","submitted":"2025-05-17 22:44:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950160-new-release","meta_rev_id":"341c57bd8be803482ad3790ba76fce865f304948","_number":950160,"virtual_id_number":950160,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:54:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:54:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5bd7a02edd59b691c497f27210874e926305d3f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a29d710fdbc9a18b02258c1f7bcb02c007798ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:19:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dbb7b1f970a4640a00c5a60a728ac8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad904def651e416fa60683cc35689d71 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cbe9481715a48148b7faa09fcd5810f : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11ef6acf29ec4faa8f4d439308a95a1d : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5e478d5dfb74c7abc34b4159bf840d7 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f4ef2b1806047fd9188f6576f372363 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3039ca63ee2445281dc46b2c568379c : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ee0ca9812af4e9db810caea995cf631 : SUCCESS in 1m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4f145cf8d4b944bba369b5251d55a59a : SUCCESS in 17m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a7770fa0a5fa40ac8e5f6d998bfbb215 : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0f4c9087cbd9483e9cf97fea42a927e8 : SUCCESS in 23m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cef3087bd3764ffb8288236a6318bf45 : SUCCESS in 19m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/daa9aaf6ed1e464d82a0f0331095e264 : SUCCESS in 10m 19s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6f6763363faa4f8ba4fb85b8867b3b1e : SUCCESS in 4m 28s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4dc3168687ea4c4faa90d56e5b7c2cb6 : SUCCESS in 4m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e015e92ff9c9430ba0dcb4f86b8e763d : SUCCESS in 6m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f547fdcc4bf045b9827c9813151362a9 : SUCCESS in 12m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0e02d124479740f6927089bf353572ab : SUCCESS in 7m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2941c05dd4654aa19fbfc5be4f947953 : SUCCESS in 7m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ecf9ade3fc454f4bb47763fa4dffd3d4 : SUCCESS in 7m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/22a095b7fe3a4df38caf4428000599bc : SUCCESS in 13m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4aaee92a84244c57a1c6024740961f9f : SUCCESS in 7m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/74a9efdc8f63414ca988990f3353b35f : SUCCESS in 7m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d9ade2133daa4961aade4bae1295cbaa : SUCCESS in 10m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3299617d896b474aa3d103c72fc384cc : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bc8e2a83c17d4547a0f797a878e88032 : SUCCESS in 10m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/db2bfb0c6ab5448b84fe3d7d6b56e950 : SUCCESS in 4m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/83c608ec4bc74a1ab661a37473e22811 : SUCCESS in 5m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8eef27a219334ef591c88f5927e000cf : SUCCESS in 10m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/466a4707338b441a94e9bca9c6c54db6 : SUCCESS in 19m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a3c47a4bb75442bbb6d2791fe6a197ef : SUCCESS in 3m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e81546675d45488ebeea00fd8ebbce17 : SUCCESS in 14m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/709b7f9419a74018a4619212a7904c87 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a3fee34493f94fc9a7b54dacc0eedadf : SUCCESS in 1m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e12722d5983f41adbb644c072d61efac : SUCCESS in 24m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d9c88618bfc845148717480e4daaebec : SUCCESS in 5m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7e7e8013a6954145974e2ff949c45de5 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4cc0fe60ae2d40d1abe8a78dff17eabe : SUCCESS in 7m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1669b2164f2848c49baf933e7bf88780 : SUCCESS in 6m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7895bfc34c1844fdaf4ddc2c87113399 : SUCCESS in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d846848b00ea4eb7bdc421d42e010f69 : SUCCESS in 1h 48m 03s","accounts_in_message":[],"_revision_number":1},{"id":"03ef74d8b173d51f7774f8b145506e6c58c0d255","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:54:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47e6423427f5f3d1a0db56761aef5d64488f5af2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:54:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"850b339d70ee09a5020f5c1dd97a3459b1408ca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d8128cfdf2463c80ea5462d0dd5267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6a4b62ad084ecd8e3f36f2b4efe311 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ca45dfc89ff43ea811f43e7a894a01b : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32571b5ddcc24f1fa33795a7124fe413 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/302035a6d8e84d228d1af759f2db1202 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/058f9eed70e04c138528de4b0bdf7055 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9169d9e251b45d6ae651abd2c248c2f : SUCCESS in 5m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7cbd86804b0f4bd38d278f45a7d59538 : SUCCESS in 17m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d9b6c16f84ec44b28df3e636a90fb125 : SUCCESS in 4m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a6453ef5337b47e5ac69e7c40f2e2890 : SUCCESS in 19m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fb1a55fb214f4a6d824f92cbab7ac407 : SUCCESS in 18m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bbf08513923242fcbe59de1e4c007349 : SUCCESS in 11m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6a86e737b48b43a09cb525b8ef586668 : SUCCESS in 4m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5fdc2d5f603547299bc9648763a812a8 : SUCCESS in 4m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/be07a90767a140108a48f8905b255789 : SUCCESS in 4m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/be6bfdd2236245d6909c626049446e60 : SUCCESS in 6m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/144dcedf286547b2b768d63a5e71ecc7 : SUCCESS in 3m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b853f35e8df1441fa41e22255c2ca541 : SUCCESS in 6m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/81a3261a55a34dfd9abd23fe000417c3 : SUCCESS in 13m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e8fea716021e42ee816a0c05360477e5 : SUCCESS in 3m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1f97a2a943b046b4b94da39b6ac89c06 : SUCCESS in 10m 50s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/37535c39ec7f4a2a861275d47e8b9664 : SUCCESS in 19m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/402eeca2d888454799256879e1b6d008 : SUCCESS in 7m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6ed5d2b06f214a2b9a1d9560117abbb5 : SUCCESS in 14m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1cfcf2cb1f80429e85131b6f8ccf4c9c : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e10a6dbec0a04119bd5a54841cbaf31a : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd4d77caa9c8423ba848bb796f7618a1 : SUCCESS in 26m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/db51f2c90a294e2ca791030c515f528b : SUCCESS in 8m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/658b389c765d43e499d34162b11c4999 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d07df55dea124a74bbe68d028b320d6c : SUCCESS in 6m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4e3b4f193b8e40ffa648fec19c7d83a7 : SUCCESS in 18m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea50009d7aa242b38f8cbcfa5bbce5b8 : SUCCESS in 59m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c83e1b11ee2487187cbc4d1111ad928 : SUCCESS in 1h 19m 20s","accounts_in_message":[],"_revision_number":1},{"id":"341c57bd8be803482ad3790ba76fce865f304948","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59b72f6c144b55df08729acc981366c48db0c6b1","revisions":{"59b72f6c144b55df08729acc981366c48db0c6b1":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:12:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/950160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/950160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/950160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/950160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/950160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/950160/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:02:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:02:02.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 10.10.0","message":"update constraint for keystonemiddleware to new release 10.10.0\n\nmeta: version: 10.10.0\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iee80c098c7ce329c00cb8cf823f74add04649f84\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic0f286bc27f6736406d54753625d9ea2e485f3ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59b72f6c144b55df08729acc981366c48db0c6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59b72f6c144b55df08729acc981366c48db0c6b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
