)]}'
{"id":"openstack%2Frequirements~950163","triplet_id":"openstack%2Frequirements~master~I036d8e1eae02c93daa5865bee4d67d8210f2102a","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-18 07:28:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I036d8e1eae02c93daa5865bee4d67d8210f2102a","subject":"update constraint for castellan to new release 5.3.0","status":"MERGED","created":"2025-05-16 13:18:03.000000000","updated":"2025-05-18 08:56:03.000000000","submitted":"2025-05-18 08:55:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950163-new-release","meta_rev_id":"ac03e1e24f2b8bef360349cd7bddd9d95c276ce1","_number":950163,"virtual_id_number":950163,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-18 08:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-18 07:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-18 07:28:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 14:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:59:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"781f89d23d84a6c732e9950fb2946f4894418cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dd72490c9c63440877ff9f62f5457f931931a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 14:19:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e399f34a54da45d78310ff560443fef2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99916df1fae431babe68495fe275d64 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0719ea90f8614fc3a81d1bd95e04a928 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/935a46b0a478499fbd49e772bd0ea66f : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/914553a3f72c49489b24f581ac7271ae : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95f0215390d3419abf826f5199a435ef : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba11144102774a4c974bb9b75a4bf286 : SUCCESS in 2m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ca9495d0de848859bf456a2788aec23 : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a47f3843839744ea84e9ecfa63fe9bca : SUCCESS in 16m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f253e84e07344f22b44896847e94639f : SUCCESS in 9m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8d123559eed94de4820f26003f1eb103 : SUCCESS in 17m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5e64c1d85e8740fd9ad71d9f54afbc06 : SUCCESS in 17m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f4d04b0c15d947a98d2f6a7508d08da6 : SUCCESS in 24m 20s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fe647f96f8e74bd1b5faacbd2ea77614 : SUCCESS in 4m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4b969d8b28f4418c80701489ef08aeaf : SUCCESS in 6m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/27c8249f3eb443cf804f0ac7775835a0 : SUCCESS in 3m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6f7535fb44ba426f8e44d4f383845f0f : SUCCESS in 12m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8987a24b11554934af9e70128173bb1e : SUCCESS in 6m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e230192cb2444b42b8dd7f627ab4425d : SUCCESS in 2m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/44ead9d5c1014acd99a40411c8c19638 : SUCCESS in 9m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bbe079e53bbd499c8531046bd6a037bb : SUCCESS in 14m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a73b20606ea4efca2936c3957ac16de : SUCCESS in 6m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6c8bc60df136493e984033f37ab629c2 : SUCCESS in 7m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6ef5a553b68c454088512a0971bc45d9 : SUCCESS in 12m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4a9540364ac646d48e81f49d6d505d47 : SUCCESS in 1m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ad37c63ca81b4e8192ca0f6e766b2da3 : SUCCESS in 10m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9e98a49de41d464092219ac01584b29d : SUCCESS in 6m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/82c03a35da084091bb1b5d4920894cce : SUCCESS in 5m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8bd7219fdc3349faa5c52a7bf9812693 : SUCCESS in 10m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3426df83edb341419025969d12e726e4 : SUCCESS in 24m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/81a42adc8e1f471a8121aa74f263c156 : SUCCESS in 7m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/64e08ab3ce2f4abca2b57bcc8cc43e7e : SUCCESS in 13m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/87b5060b0a5249b693ff83c462869fc7 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6742250396c4bd7be187d0369508c2e : SUCCESS in 1m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e322f219e3a4906b6f31e45a2139f17 : SUCCESS in 26m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5af1c020c19a452384896f23df0aa9ad : SUCCESS in 6m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/644d56b011304f2381f85d78aa362066 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/853b87104b714260b52cf1e37a6de82f : SUCCESS in 6m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cfe06563504845b88305efedcac06bea : SUCCESS in 17m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d957d144c308464497773fb903504f60 : SUCCESS in 55m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a654a0867af744f594d3fdedba8f1803 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4708b6951cbe541ab22a281087d6d187e195c5c7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:59:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebb02c8d8d848dc1196a75bfe7eb70880d96d584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:45:01.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 950163,1","accounts_in_message":[],"_revision_number":1},{"id":"fcf7550baf448d9ecd95368c40af94860eefa22b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 04:44:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2542c04c74bbe97d014814af32752e501eb2e7a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 06:03:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c8c622e98f54212bd28355ff2ee273a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a66e17ab3e4a49bea3bfea2bc911a7 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64bff3dbe90e4faeba56dcd931dba10e : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bd85de4200a443e9e32120758402b90 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5ad83ebf3da4177a03158fa43628797 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc6981c874014ceb8842bb21374670ec : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ce896619f134d76ae7e19d23f45856a : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4e21c4176ae4520bb40514902d5ef24 : SUCCESS in 4m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/df7a1caa6e414430967c8415ad1fbe63 : SUCCESS in 13m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/17f7887592e048b98d0f20721d46a4a8 : SUCCESS in 11m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4ab9e5236aa84e3c944ca23b3fdbf1b8 : SUCCESS in 13m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1e87ce906bd94226ae9faffce0205a97 : SUCCESS in 8m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d0e5dff2a3ba43ec8425099174e03667 : SUCCESS in 9m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fac4b0ae06734c698e1a0bc9302a1997 : SUCCESS in 2m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/48acab5679854d86b17269ca517bf22c : SUCCESS in 2m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e0c13d34916141ba8b1b62ae4b391d26 : SUCCESS in 2m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8efd6b564a6a4b959cfea9c4f1e661cc : SUCCESS in 11m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1e592f85572649bea4b9529e81102a76 : SUCCESS in 2m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9d75ecc3ac3f4bb580db6ec05ae83759 : SUCCESS in 5m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ebe3c878de9a45338c893298e7bac20c : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a40981e63776430ca227b4a35c4d48be : SUCCESS in 6m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5b1e732cc714d69887e6e63a2a342e8 : SUCCESS in 6m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b05415c3f8484c9b82918e0ab5f6556d : SUCCESS in 4m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/77c2c016967e434a8a6e18a07bdb2e70 : SUCCESS in 9m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dc75241ebf99421a8ab0ce740fa7e3ac : SUCCESS in 1m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/31a2712599ed45cfa9594554b702a669 : SUCCESS in 10m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/00924ca8677343398640a4a5fc8892ff : SUCCESS in 2m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/019087e6375d44e0883dd59e26784b83 : SUCCESS in 3m 14s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c9db5bc3a5274b46a1c456ffa1c9d54d : SUCCESS in 11m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e295180ce0fa456b84cf193eed476d80 : SUCCESS in 15m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5109ed1e148c497ba34d5800b44c0a57 : SUCCESS in 7m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/405f8bc51e48423383d51ab696b682cd : SUCCESS in 10m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4fc1ef0370c941ebb52c4a13b293171a : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0102c4f881874802bc126f84e1d0d874 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a5c721e69604c55b74025ccce2d693b : SUCCESS in 19m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/89091c26454a45d0b7843d966be81423 : SUCCESS in 7m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c911c924525e4b36b5e3997a49ad71c9 : SUCCESS in 3m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d82a6a3402fe4894a03d97280372836a : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/72c1aeaab24c43b8810648d59c072446 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab54207549f644f2b6481b8f0b5ba701 : SUCCESS in 44m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f39a7f9334944086b30a4e40e43d9e96 : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":2},{"id":"838a8025bc85669ba4c76260159510167d9ef9f3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 07:28:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f00165aca2bf12e8c58841f7db2ad4c2831e1bb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 07:29:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be13f6457ee6f411a8d307f488c5aee32050771b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:55:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e92be56ac941c7a3b5853bd3f3c175\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed00f4b8c0ba40478b3b14c0f85f4367 : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1363a2da43402b96916a701d4d5c71 : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f867821f602406783fd0d8da24c78a3 : SUCCESS in 1m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc008f2e534f4e6eaa84545c3cca7bdd : SUCCESS in 2m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b27b60fbf794e8ba9c24f5f2dbf026e : SUCCESS in 2m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9bc7ab0802b427d806c406b6f590bbd : SUCCESS in 2m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/49995b7146e74649b4885bb162d4aee6 : SUCCESS in 6m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/66d6179dd6634ba6bef0267f914dd1e0 : SUCCESS in 7m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/92272c880b9049fd8727953020676fce : SUCCESS in 13m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/936395789b77433eb7b98fa523fcedcf : SUCCESS in 9m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9a7d9c73634347f18adcea1404e1a855 : SUCCESS in 11m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/592e03a58b3d4589a712a3a61ecd882e : SUCCESS in 4m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8e60d26565f94d78b581c514702708bd : SUCCESS in 4m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f3196f7aa4624874a64e7f1e6f6957c4 : SUCCESS in 5m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c595f1023af144e3af19aa1d26f1cacd : SUCCESS in 13m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22dcd377a9fe4b2e957424880b6dd1e6 : SUCCESS in 8m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c244bf700b9f418bab109d808d026f84 : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/35a31be1b60b43d1a04a563fd9c12c76 : SUCCESS in 11m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/beb58d98056b46dcb37499ce3b25449d : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d6132883a6f94bc5ab2afea71fc8457e : SUCCESS in 10m 25s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3d8c6a17218d49019f6fb07acc0af916 : SUCCESS in 22m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2af5b9df45b64247a8dc9c59b99ec2af : SUCCESS in 7m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c0fb9a13d50a4d15a3bad2cd7a539f5c : SUCCESS in 7m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ba5820c57ee24fc5a95252035f423a88 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32830e8fcf7043a2ba4f6f1d63d9f7a2 : SUCCESS in 2m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd296ee8b3b448118dd8ac744ae09f28 : SUCCESS in 20m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/46efac4a259b4bfc97ee6b797d335aee : SUCCESS in 6m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/251d776a19d4463d8a79d3deac752674 : SUCCESS in 3m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53adfb580ccf42bf944d50326e4e0a33 : SUCCESS in 3m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/11b757b0ec2d4d849f0eb6aeb72b3ba0 : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49108c12a17a47a382f8376b237ebd71 : SUCCESS in 1h 01m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2082891a4964c8b972b59f5eb30d3bf : SUCCESS in 1h 25m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2a30029a092698501d6cb71d039e925a16c38d46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:55:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ac03e1e24f2b8bef360349cd7bddd9d95c276ce1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:56:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/789a3d1a91d34bb895e857e0cb254893\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5d50893c64441da4900c7fd9b75d97 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b3255c2a223523583d0b6214e7c092eae185db9","revisions":{"e92f12a31d475f33838298e3557b7ddc7ca4c382":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:18:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/950163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/950163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/950163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/950163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/950163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/950163/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:08:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:08:02.000000000","tz":0},"subject":"update constraint for castellan to new release 5.3.0","message":"update constraint for castellan to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7f53e90fd5b0f49c4f00d054666ff7afd9943768\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I036d8e1eae02c93daa5865bee4d67d8210f2102a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e92f12a31d475f33838298e3557b7ddc7ca4c382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e92f12a31d475f33838298e3557b7ddc7ca4c382"}]},"branch":"refs/heads/master"},"5b3255c2a223523583d0b6214e7c092eae185db9":{"kind":"REWORK","_number":2,"created":"2025-05-18 04:44:24.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/950163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/950163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/950163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/950163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/950163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/950163/2"}}},"commit":{"parents":[{"commit":"5563c94f0716fb87845fa84255766ba3d522b0e0","subject":"Merge \"update constraint for keystoneauth1 to new release 5.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5563c94f0716fb87845fa84255766ba3d522b0e0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:08:02.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-18 04:43:17.000000000","tz":120},"subject":"update constraint for castellan to new release 5.3.0","message":"update constraint for castellan to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7f53e90fd5b0f49c4f00d054666ff7afd9943768\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I036d8e1eae02c93daa5865bee4d67d8210f2102a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b3255c2a223523583d0b6214e7c092eae185db9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b3255c2a223523583d0b6214e7c092eae185db9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
