)]}'
{"id":"openstack%2Frequirements~950164","triplet_id":"openstack%2Frequirements~master~I68605b51337c91cae90c43e0f9f31e663841fd0e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I68605b51337c91cae90c43e0f9f31e663841fd0e","subject":"update constraint for oslo.config to new release 9.8.0","status":"MERGED","created":"2025-05-16 13:21:20.000000000","updated":"2025-05-17 22:44:49.000000000","submitted":"2025-05-17 22:44:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950164-new-release","meta_rev_id":"a3a20dd11e5679776a38a70005777d8703d997dd","_number":950164,"virtual_id_number":950164,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:56:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:05:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:56:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0611cd14abd48bf9d1eed818440fe161ea46dded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:21:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7af3eafb5fb424a98ded38944577e1828395775b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:05:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc5376060fb4915a175158e81630eef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cbb4dce018408abc679bd50983d521 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b9765d33904bac93799389c91f6564 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c20bb9977cd14612aa97e7ca47f46846 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2dedb1d09fd04610be9ccf3e48172cc7 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a3ba28e6dc34e97bc56dde51baa2233 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca47ae6650ee41d69def84a7cb99443f : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/295c27dd355d4b60a7f2c3bff87df20d : SUCCESS in 3m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9f0d3ea131a14f27b0415ccd3c9917b5 : SUCCESS in 8m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5d947bec99204cf88f3234b40cfca998 : SUCCESS in 8m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ceb241ab67214c088604a1690a3f370f : SUCCESS in 15m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6b22c368cb8943f291da51ae0e4e8089 : SUCCESS in 23m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/60c4d9d8ed3540df8b2b3d14c0ddce6e : SUCCESS in 28m 23s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/360e278f9e8549a3b7d98e8a966ecb08 : SUCCESS in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b2dffa7b690c43d6a96580514070620c : SUCCESS in 2m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b0235e917b7d46d0abc85dbc389d6a0f : SUCCESS in 4m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/907a939ca4b54be0beb6723e4611314a : SUCCESS in 12m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/eeaf9cce90274c30b653d6ede993cf3f : SUCCESS in 3m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d8d00c6405a4491a9cd4ace92f2c1974 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8ef24be2a49342048ff7189ff20a3b09 : SUCCESS in 9m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1344d0eba6d24b37ab1c64a1e4c2e3fa : SUCCESS in 13m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/32b4b789d4a14059b36af33f3d0fd457 : SUCCESS in 8m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4441e7c0ded149f28437b9381f348709 : SUCCESS in 8m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4f52885ad0304baeaf49fe9227095cfb : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2f4bc0f22f714127857c9f12152e20d7 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/69b3eb2c28984d088763789381b84c62 : SUCCESS in 8m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0d137076ca7e4b59857aeb834bc2a23e : SUCCESS in 2m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d3ac6116ef5143bb8b77e79bf744655f : SUCCESS in 6m 13s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cc7c5cdd37e643adaf9d65c98c57c7e9 : SUCCESS in 12m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/47a96d358f3242699cf16fa080984b29 : SUCCESS in 19m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8a4fd0c4bd5649e89bef90b73ada1247 : SUCCESS in 8m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c30ded88f8104849a76700c447c95fdd : SUCCESS in 15m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a90c320dccc34a5791af90abf23d1c95 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52d0b21f3a9d43fca332a788012b77e0 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ead486fa3e9e43bfab07c6f539dff842 : SUCCESS in 24m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/15a6546a79b24c3691146841af170a98 : SUCCESS in 3m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c9f238271bd44fda90de4e372a550c98 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc72ffb2e5914a18884aa19de3b447b3 : SUCCESS in 6m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bd931024841a44a79e424b21defdd462 : SUCCESS in 18m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf9cf88887c243608be8f0e614b049bd : SUCCESS in 54m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfe2d8488aa64a78ba2ebf31f95110f7 : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3ad6210c9acc4f7daf8cf23de1b110cc3043d8fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:56:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73399a6b3d887080c1e9382348a6551a1f3d4a18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:56:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbf28bd32dab2d5a2f443ce1089b3e0f6669d6b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad03f84010e4438f9cc58953fcab1612\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a38c95cf89f46c48dc2fceef93f95ac : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ab374f0e1847dc879358dd54f1b829 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc11d114a17c4dfbb7acfd9e1876e0d2 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76a0d0af7bbb4341875cdd364edf6f0b : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2565040b888b4eb7a81eff53bb5fa637 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/336e8df20b314c128eb5af1c187b1e96 : SUCCESS in 2m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f8338b1efa544f01b63f88b10e375710 : SUCCESS in 9m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/224a7aff930e493da626645d4fd39d07 : SUCCESS in 4m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31d6304b9abe476d899527b7abab81c1 : SUCCESS in 17m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2864aaad26bb4389b4e9879400033308 : SUCCESS in 23m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a5055f165a7e4901ac5a29791fb57568 : SUCCESS in 12m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a9682ab79bf6428b83dd13244ca7dc38 : SUCCESS in 6m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/42a0c68006324575a068953b162972f1 : SUCCESS in 6m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/db05214b3f58420f83a251ff31d8c0a5 : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/00900f71cdb749548eb0258b72981f17 : SUCCESS in 5m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c185b3d8352468ab9db639b3d5012c5 : SUCCESS in 8m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/01127595104247f5b11f0d32258655a8 : SUCCESS in 8m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f271cf6818d246f6ab303cc6c3b01d69 : SUCCESS in 11m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/931061c01678470981e9d4f7ff4c6fa8 : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/67324606c77e4c00ab3968116747ff1b : SUCCESS in 9m 25s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4407b1d55a8d412ab20fbf5b41a554ba : SUCCESS in 20m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/df8da1f7fe8b4cc2b109ffec91a5068b : SUCCESS in 7m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9e46d1e18e294dc89574d3176a21f2d4 : SUCCESS in 8m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7133412746db40229b41d54ced13562b : SUCCESS in 6m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79703c0adef941e184bd74648a940a4d : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9be00ca1409a4c338ed4dffd8b40c41a : SUCCESS in 25m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/51aee0e1437a4c6294ef43f9cf5adf1e : SUCCESS in 6m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/47993246f0e7423b80266216c7e26800 : SUCCESS in 3m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cfe500c8a1646fca05ab6f4aa1d4c64 : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/17cf4cfd3d1746489d673cf6967b1cff : SUCCESS in 15m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f70954a493c4448c93efb46635ab0a81 : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b593b70965dc4911a6cf46795823c889 : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a3a20dd11e5679776a38a70005777d8703d997dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21347646b06d9d21ecb01aaca9e3bd90aa5e0fbc","revisions":{"21347646b06d9d21ecb01aaca9e3bd90aa5e0fbc":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:21:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/950164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/950164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/950164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/950164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/950164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/950164/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:11:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:11:18.000000000","tz":0},"subject":"update constraint for oslo.config to new release 9.8.0","message":"update constraint for oslo.config to new release 9.8.0\n\nmeta: version: 9.8.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie8fd0859c1ad106acc23f1296d07be2a8e6051ec\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I68605b51337c91cae90c43e0f9f31e663841fd0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21347646b06d9d21ecb01aaca9e3bd90aa5e0fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21347646b06d9d21ecb01aaca9e3bd90aa5e0fbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
