)]}'
{"id":"openstack%2Frequirements~950165","triplet_id":"openstack%2Frequirements~master~I190e4f9f235a09fe57e007849fd5aa588f2e2d0f","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-19 14:38:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I190e4f9f235a09fe57e007849fd5aa588f2e2d0f","subject":"update constraint for oslo.context to new release 6.0.0","status":"MERGED","created":"2025-05-16 13:23:22.000000000","updated":"2025-05-19 14:39:46.000000000","submitted":"2025-05-19 14:38:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950165-new-release","meta_rev_id":"04b33403b58618a6e97dcc35baab3439d83ed068","_number":950165,"virtual_id_number":950165,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 14:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-05-19 12:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-05-19 12:58:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 15:13:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-16 15:13:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-05-19 12:58:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a1abd099c80cc03340c15cc3191c4a694e85d41d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69a81501c6c9e76c746a611aef3cd434601af811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:09:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1be9f3475dde4464a9f923d346c29281\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f41296fc7034e7eaf72ced6251abb22 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b550dcdd56804304a07b2c7179a8856b : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee27fad38de54ced9c6e974ef8873270 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/828d209eddbc4783b24ed45df0b37dd0 : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7726f78e41e7431d97f1aa81518ced67 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfe1d9251a9d47dba6f3a556f8d36c2a : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eaf553535502482e9750706b4065df2e : SUCCESS in 2m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ed9e7e28ab1540b49a4b3131446710c9 : SUCCESS in 17m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0b214998b2cf4cde98aac0fd20091d7e : SUCCESS in 10m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3411c2099a00457eba18cb95a125297f : SUCCESS in 20m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ddedf681e0a0446c81abcd1a6a4bfefa : SUCCESS in 20m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/57439f67acff4f6c914e54717169a030 : SUCCESS in 24m 32s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/23e30ed2b520442485a842b5603bc8b2 : SUCCESS in 6m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2cf7ec95ab4d472c94d1c25793a622cb : SUCCESS in 4m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b350f97c4d064d87a6f909a24109f46e : SUCCESS in 4m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/76f42aff40a0424ba5f82ad1d2662d0f : SUCCESS in 13m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/25cf6d43ff9448778bf7f3daef4a7ede : SUCCESS in 5m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/44d20b0cea774850ac9cd20ace523898 : SUCCESS in 6m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0982a01a0f1c4762a51bc2676a15d5c1 : FAILURE in 9m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5f9661566b774778a74e63c66086a9d3 : SUCCESS in 14m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3da010d7faae43849ff4f553c58dd1de : SUCCESS in 7m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7968257121a84ef3b43f8b8ddd62d803 : SUCCESS in 7m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ace86dea3aad49d6acc3774c7875ad71 : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4c400bdd4a6949ba87c7c28ea488c57f : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/02b56260728c4157a90a8ce09b4bab5d : SUCCESS in 10m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ceab62cec6744a99861c61b71b5580c9 : SUCCESS in 2m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4edab0d2c4a04628a588ca29db43a0d3 : SUCCESS in 4m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2d7df05cdc3d40069b04fb2fffcd53a7 : SUCCESS in 11m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0b9aa7afc16c4b77a8cf6526a312abb7 : SUCCESS in 19m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/16d8e76a103f43c79376f5f62d3ad340 : SUCCESS in 7m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f039bb98e4314ecca998bc0486671b49 : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f6c1015303ba413aae61b90927e1da67 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a491c08fa3b4e0fa60b9cc710eaaf77 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fba5e37c9f704337a12d9ea2b7fc4ecb : SUCCESS in 28m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1cbe9682f15344cd8a68e180803ef930 : SUCCESS in 7m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d86cbb797e38416cbd27762e9ef83786 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf3dc524624348d9837a293c8bc0f8b9 : SUCCESS in 4m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a9d2a61d1d44459387a7627f2f2b64da : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1f00989e9014ac38a7aff16132ee5ef : SUCCESS in 1h 04m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/596e5b84ac6d4403b35a12e7cb7a30ee : SUCCESS in 1h 40m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e5b7c23e16dd5e9376cf28f413ab533860942ce7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-16 15:13:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7f953e0c88675c630ce4f19a68af564097ab06b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-19 05:32:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fed7e3a1ae8b037fd2c93a2268d91bad22abcb9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 07:12:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a0dae4bf67f45489a2739fbcf1ed436\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd83f1be91e4485ae1a8e2d24ec5955 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce409d59fc54965b884795946630541 : SUCCESS in 2m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b19cbe0d196246b3b93c6984620d7e48 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a70f0df3b4e46af89bbdd8ee7d90fd8 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49b39753af484ee4aa0d9be3c11c74ae : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2743306e65914f0fba371dc3fe31bd48 : SUCCESS in 2m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/548d682754ef492388e09ef0e2fc0eb1 : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/286376fdc4ee401386a5bee7dbe1119a : SUCCESS in 6m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/faaddf9ba57e4212b8800c7a78e12b68 : SUCCESS in 6m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2bf275f34aef4bd5855d2940b129f448 : SUCCESS in 12m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1b583bd20a214a3f8709d21a98e24521 : SUCCESS in 13m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fc1dcd97152949249120a794c835b741 : SUCCESS in 10m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c89a48dacc124922a3346f54d56bb3cf : SUCCESS in 2m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/61cf2d2fa2934cfd8531bd758f153992 : SUCCESS in 3m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c876a28d831f4dc4a4b649d4c74e481a : SUCCESS in 3m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9ab2c739ef244a308f34fd0864e8eadc : SUCCESS in 11m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d3456858f5f84b7aa871ae552b29f888 : SUCCESS in 2m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5f93a26b266e4421876f774869955834 : SUCCESS in 4m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a49d58960b184bc4ac1c7ab376e58d37 : SUCCESS in 6m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fdcf024c17f3465ca10efcaf565c941d : SUCCESS in 5m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d479a6a2f6e0495ca805330e3186d84f : SUCCESS in 8m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1a4dfe8629bf4a759f217c626b62f672 : SUCCESS in 6m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c09882704a9a456d86d1fdf6e076c6fa : SUCCESS in 11m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dc675b7d7b074683aedf266a81a2af85 : SUCCESS in 1m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7126d5629163423ba870ef85eede82e1 : SUCCESS in 8m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/23a94040fcd94df19aedd241eb5df046 : SUCCESS in 3m 29s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7d7febe59ae04b54a79aaf3b3033f8a6 : SUCCESS in 3m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4061a92fe6c24c2985abf50415f2f4d9 : SUCCESS in 11m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e43f17bd99064cd8bdd10d1ce6f66a62 : SUCCESS in 15m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3bbdcaf0e52a49d6ab2c384c8cba6a04 : SUCCESS in 6m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/16ece035ff9d481cbf510b7a6e228add : SUCCESS in 9m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/376721cd92f840f981623244d44e1a00 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8fde74198b74aa78a2f1a3753218be9 : SUCCESS in 2m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a735f7926bdb4971871e94c2a75ead2e : SUCCESS in 25m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/636819265a0741a3a26ea49a5980997a : SUCCESS in 4m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f7722e4249284e21b78251dbf59a6e0e : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa33f332552a4cb39b5378231a8c9403 : SUCCESS in 3m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/153cd8a4190a4dcaae8becf0e857e08b : SUCCESS in 17m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb7a6ff8323c41c7a1ef02a43da97005 : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4e6adaf505744598bd1825e148120fe : SUCCESS in 1h 39m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6d966b2c075552b2e0e0f66e4673d38b7960abb3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-19 12:58:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb667f9385ff3f7faff593bac3fffc5005873d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 12:58:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f080d27f5668233dbff1efa39712f976f0b91171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 14:38:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3265f24cd4434a699add3193b0f12479\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feac0eebabd447debe230cb2a1c17a85 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac405f7fc016407aa4db110a76b72542 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/557b44711eef4b309f8bd15d9a3f8952 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5a7f35b1a0d481797834d19bbb30218 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/818fd60e32964b529d9c6c7ca18fad1a : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d008ed2af4348fdac6dbbe60f0b6b60 : SUCCESS in 5m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6a8bfc5ee8064050b842bf9ae9c34b87 : SUCCESS in 20m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dd38119e88b0420a9c813b2191663386 : SUCCESS in 12m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/26813699149242dfae26766dc3ba0b98 : SUCCESS in 22m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/11608eb085c04f8b8ab2442dddda7188 : SUCCESS in 23m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e333b93cddf744c78d124d9f5053b5de : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6a91584b6df84a9988c2c9201680cb4e : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d2c7a9bcdd0547f291ff5134528fd2cb : SUCCESS in 8m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/da544ab64d9a4a03a7ea9e62aa949230 : SUCCESS in 7m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3f9f8a902dc947b693681ab627e8f728 : SUCCESS in 13m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc5469ff50c34866be9fb83b61d929ce : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9807d2935c56452981ec7fec2b218138 : SUCCESS in 7m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a620466ad1284b029c6a69343f9dfb64 : SUCCESS in 12m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a379aa52e84148df97c0578b6d76088f : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/be329bb8ef1d41adaa2aa0c1d974b690 : SUCCESS in 13m 03s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f89040403cff456592681aaeb03219ab : SUCCESS in 19m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e3a17ee7eb4c4905a27e7cffc0de4127 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a2f634769ab24dd08154128cf8a045b1 : SUCCESS in 13m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/020bd413b47d4de5a224e903b8e237e7 : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e61b3b518b9e4e7a86b25723a61bfc99 : SUCCESS in 2m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28ea926b169b403383b2c414258f52ca : SUCCESS in 23m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/076a2bddee574dc4b5589ce6cd80c9a4 : SUCCESS in 8m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/09d0142650574dc1b25be0c3ac1ea4f7 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f34da4ed402f43178d8f94035d64ba7f : SUCCESS in 6m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/11dd9403ab354e42ade091225dbb31c3 : SUCCESS in 15m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27fa1944e9f34e1fb03da2b87983bce4 : SUCCESS in 58m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0245d97afe6b4078bf244bda45de0381 : SUCCESS in 1h 34m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3f5a64ab23eb147b87748a80f2104ca443059f43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 14:38:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"04b33403b58618a6e97dcc35baab3439d83ed068","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 14:39:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c7125a04654b0dbb76a3f62a5b62f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd40f42d478f4545a075b7d4958695dc : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55e5a14386202c4e4baf88598a4fdc5b9a0b677e","revisions":{"55e5a14386202c4e4baf88598a4fdc5b9a0b677e":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:23:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/950165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/950165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/950165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/950165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/950165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/950165/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:13:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:13:15.000000000","tz":0},"subject":"update constraint for oslo.context to new release 6.0.0","message":"update constraint for oslo.context to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I6319fa8285cfb7ed1010c1e03e301e18bed115f8\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I190e4f9f235a09fe57e007849fd5aa588f2e2d0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e5a14386202c4e4baf88598a4fdc5b9a0b677e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e5a14386202c4e4baf88598a4fdc5b9a0b677e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
