)]}'
{"id":"openstack%2Frequirements~950167","triplet_id":"openstack%2Frequirements~master~I374a5ce5fec9696ffedd65bb13627ff19ff1d183","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I374a5ce5fec9696ffedd65bb13627ff19ff1d183","subject":"update constraint for cliff to new release 4.10.0","status":"MERGED","created":"2025-05-16 13:34:36.000000000","updated":"2025-05-18 14:09:28.000000000","submitted":"2025-05-18 14:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950167-new-release","meta_rev_id":"427b4a5bc25fd7bf3c6c6c71705dd22cc3512717","_number":950167,"virtual_id_number":950167,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-18 14:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-18 12:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-18 12:28:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 14:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 15:33:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-18 12:28:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"217cd38c02eee6da01f03a2d9c581ff2fd44911d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:34:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01d6596d144353d97e3a015b718feeadabd9453f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 14:39:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c496f20d4c14d6e811a25a7d00391c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d686935c2c604e63abed4e4680c17edf : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f1206dd5c684c5f9b56a20a233de9b9 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dd43df17b6247439193e6f2873a01b1 : SUCCESS in 2m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b244569178d46d592ff313e76fe0143 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/053ee46dcdaf4bd4bc613789cc8fcd2a : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffe225175eb446429c91b60cd93a0003 : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3c94ffe4da14ba3b00177640c1e2c5c : SUCCESS in 2m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/04d7b6f2d4da47baaaab4446ddb708b4 : FAILURE in 2m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8211a3e16d8d46e5a33201878c5e29dd : SUCCESS in 10m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a16d3c5aba004ddcb955e8ca5ae0f266 : SUCCESS in 16m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e59ed91fce6c433a9f46e034d81b4db1 : SUCCESS in 24m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/408f4edabb8d4ff7a6fb35bf60e1f06e : SUCCESS in 30m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/adc7bbae338e44fea139a68ce6eb76ca : SUCCESS in 4m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/28ed6b7acb5d4d249832a88f5cbd2b97 : SUCCESS in 4m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2bc68ad76d794496952131ab5e84a47f : SUCCESS in 4m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/59e58d22006d4815b65c272d1fede203 : SUCCESS in 14m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/db69e539c9094ec98a84e6858e9196ce : SUCCESS in 4m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c88fcce17f7f4427aa4b07231bc4f25f : SUCCESS in 6m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cd1ac30060544f688429ab8dd6fa156a : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8d3682ecc0104aea83def7979e61b3c2 : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0026e80fae4a4ff0877b87a61487f9c9 : SUCCESS in 5m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/576ffc4473764736a8a3b7fc5889d753 : SUCCESS in 8m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3d2bd85c5b8d47758dec932a6deeb68c : SUCCESS in 11m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1555478b6024489ca81618f941343aa6 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0093de3695f1464a8a4068a8d6576faf : SUCCESS in 13m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/020eac0772c446ea858f15e9c62fe630 : SUCCESS in 2m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/47e9fad1b4344bc8b60de23a44f3a68e : SUCCESS in 2m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c1805680a4fa4937ace203d404e47a3e : SUCCESS in 15m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cb887d52886f439c99c8fcab1f2f8ee0 : SUCCESS in 25m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2defca5d21154e4c931357a9daac3d60 : SUCCESS in 7m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3e314da2ddcc4ba683d9da7852d21c26 : SUCCESS in 15m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0686da9a7bca47c9916dbe0d47351fbe : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/97a3067e850f4b61b494f6db5a644cce : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa022d98c8134216b1ca4502b5c31e31 : SUCCESS in 27m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/83be5695486e4c1e90bbe12e63e62c4e : SUCCESS in 8m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7e22cae2ad70461986f0ed0021fb7fcc : SUCCESS in 3m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45fb4e85137847059b096e9c69841c70 : SUCCESS in 3m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6d57b4e42c24478d9cffb703c9e04e35 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e653299bfc44033b2d6598f688f701c : SUCCESS in 59m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b275cf6182a24bcabcd569b79b7e5d2d : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0f168aab288cf61986e8336c9c73e70c45a664c7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-16 15:33:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83f44a5f812d13b207161ba73d786319d2f006bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 07:34:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5f7c90c9d9faae51b3a87762b6a48abffff72ac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 08:56:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa00c633b054a83af6cad1864b64c64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce12d1c7d0349fca03a020fa554f129 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad0d2e0854804bbb9e80c0294b969890 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc86f8b369324113bbf343460b9ca93e : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3666fbba2f4d4f7b82936b53d799f192 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3d30b60276e4671b7c354c40fac73e9 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f818d04afc594fd3b493573ca81f55e6 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05ebb199dcba416c8d2225c85c5cbe6f : SUCCESS in 4m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f0f33af9bb0e48f0ba819451a9771392 : SUCCESS in 17m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7dbe1c19be09421887d45ea230ac35f4 : SUCCESS in 13m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b8df34e65fb14fa2a9089014d1edec3e : SUCCESS in 20m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6c36ed1799924a2898faafeb1e99c267 : SUCCESS in 25m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/68b3b08b130542d49ff781f5782db8a6 : SUCCESS in 9m 45s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fa034f8c9aa147a98d440376d4e74187 : SUCCESS in 6m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a48efb40c1154f4191a8fa0c85768fd5 : SUCCESS in 4m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e2f6cf74aa294389a9a38444fab4e9d9 : SUCCESS in 2m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/38f2ffedf7384506a7f7997b1f0d1e3d : SUCCESS in 9m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/785612a5b1bd41aba1027d347cfec885 : SUCCESS in 3m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2092760e16174417ba0f1bf96f29ef9f : SUCCESS in 2m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/78afff2d4bdd4a39a20ff0af6cafcc70 : SUCCESS in 5m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f9413d1748294402a5a1884e392740ab : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7c6279ad94e4a5d84d0a95064c21b77 : SUCCESS in 5m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0cc7c134ecfd410ea0f55506d1305a1d : SUCCESS in 7m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5725b5d8e9d14d06a568905aa501cfef : SUCCESS in 10m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4e4d8d1406944cdca6ebecd71a7270db : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/36797c9151594c5298df136c10cb3244 : SUCCESS in 10m 42s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/518e79fdd2b44c1e900858672c516def : SUCCESS in 3m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7fdbe064d34b49d88baa0b8b28ad14a2 : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f69c92be2b3b4da788dde4584b0ff903 : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d8dd71601307444b9db2f035a99996c6 : SUCCESS in 20m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8729ca78569b408b938e1a5f8fc2e543 : SUCCESS in 7m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/de65875d112a488788e037ea109abecf : SUCCESS in 13m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b4693fceed30458b9700d7bd25fbfccb : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/836cff3c5e2744c48207d6b95dc00cc0 : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59c4253973a742e5ae89dab95ecc7414 : SUCCESS in 31m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c37011411acc4a56a4b79cd7f4e12ad6 : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/42723dac0ebb40b380fb6c5c1776cad0 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03575a0c7ac2472e93a42132bd699809 : SUCCESS in 7m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f52710333046447d9e51e86992260f4a : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/601d7862565646d6b6a568a1e7f34ae3 : SUCCESS in 58m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1bd3fdd85b24b4db418991dfc659077 : SUCCESS in 1h 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a15749bd13ec9bcd203d15c84ff1c76ee8d4a969","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 12:28:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e1d7c5677d3193e669aba88b2a11146aa602949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 12:28:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"882393050d9422e8b2405a6c7796a1abb675bcb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 14:08:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da5726907eb44d1a46cedc1da82f472\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1ad9ce619a4a87855b7de34d62cb72 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33dbf766d4cf4fc4892f1b8754a100f2 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2250634c697e402fb3524454e22afe40 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24d5c9fd219b4eb98bedb3312902ef50 : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9e963375dcb4502a33906a96ebebfc7 : SUCCESS in 1m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff1c2a5d5ec540b0893a94a0968a20ff : SUCCESS in 1m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/62284aac39094a8b90ea4c841f3c6e19 : SUCCESS in 9m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e65d49f5ca043a5a1a29e5110e7e332 : SUCCESS in 4m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b52027699c4f40a194292cfec55df7d6 : SUCCESS in 7m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8465b8165a0949369c1e2bbc8cf67255 : SUCCESS in 10m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/423d156e9e684cda83571e156cbaecc7 : SUCCESS in 6m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/979827720bb04dec8cf3bb0ad27feca6 : SUCCESS in 3m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1621398ba70340ccaf221f3a3582a0b2 : SUCCESS in 5m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3b5de2380dbf46a4b13815f6bfa81e41 : SUCCESS in 4m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0c63d3fd364b4e90905d15a1fb737ef5 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54d451b6abe44f72bc53877e9f13ebc0 : SUCCESS in 8m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7e676cff094d46058772c078de385c29 : SUCCESS in 6m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4b2b3458dce14070bc85b6fb77375b47 : SUCCESS in 11m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7f0fbc1d53bb429da4a76409e4ef6985 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ce7ffbd5540f4493b25cc9c516ae1afe : SUCCESS in 7m 58s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c36e11a6f86049bcb2accedf0eea8a59 : SUCCESS in 17m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d48c874275c4bf5997990baaac4324f : SUCCESS in 7m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/70cd3e7d7118418fbf09df8c1f36f4a7 : SUCCESS in 11m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7c845b70ccac4385b900e225b421280d : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8753a2f5af7f4791bcd355370491cb3d : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/986d42a58ffa419d97bb5227d50ab1ca : SUCCESS in 21m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bdcd812300684b6ba211e6719ba89148 : SUCCESS in 6m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d4a0982dc77545d58a29653ce6682c7b : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e078063df284bfa9075abc3a0babecc : SUCCESS in 5m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/09b6d380cf8f44679c309e684706776a : SUCCESS in 19m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61461be7064a4044b3f1fe9d8651479b : SUCCESS in 46m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d77d79b8d8ce4359b138077bc0c59fbf : SUCCESS in 1h 39m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9d3219c7cfd4326522b86b03148c4e1474e8b3cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 14:08:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"427b4a5bc25fd7bf3c6c6c71705dd22cc3512717","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 14:09:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c679dad8f91b4971b0b725cd6d0d8464\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d8d965f9d64ebe8fc434780e85faa9 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e99201c91c4dbe12cd93470d92279e53190431ff","revisions":{"daf47ff9f9bc1240999195866f83c23288de61f0":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:34:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/950167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/950167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/950167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/950167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/950167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/950167/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:24:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:24:35.000000000","tz":0},"subject":"update constraint for cliff to new release 4.10.0","message":"update constraint for cliff to new release 4.10.0\n\nmeta: version: 4.10.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9bb78e3066be6332ffe7bdee48e85a6df6fb5fc6\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I374a5ce5fec9696ffedd65bb13627ff19ff1d183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daf47ff9f9bc1240999195866f83c23288de61f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daf47ff9f9bc1240999195866f83c23288de61f0"}]},"branch":"refs/heads/master"},"e99201c91c4dbe12cd93470d92279e53190431ff":{"kind":"REWORK","_number":2,"created":"2025-05-18 07:34:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/67/950167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/950167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/950167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/950167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/950167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/950167/2"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:24:35.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-18 07:32:56.000000000","tz":120},"subject":"update constraint for cliff to new release 4.10.0","message":"update constraint for cliff to new release 4.10.0\n\nManually amended to bump the version only for python \u003e\u003d3.10 since cliff\ndropped support for py3.9.\n\nmeta: version: 4.10.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9bb78e3066be6332ffe7bdee48e85a6df6fb5fc6\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I374a5ce5fec9696ffedd65bb13627ff19ff1d183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e99201c91c4dbe12cd93470d92279e53190431ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e99201c91c4dbe12cd93470d92279e53190431ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
