)]}'
{"id":"openstack%2Frequirements~950168","triplet_id":"openstack%2Frequirements~master~I63656996e058cd7c8ab4f49e1529df2ec0d8eba3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I63656996e058cd7c8ab4f49e1529df2ec0d8eba3","subject":"update constraint for python-neutronclient to new release 11.5.0","status":"MERGED","created":"2025-05-16 13:35:34.000000000","updated":"2025-05-17 22:32:28.000000000","submitted":"2025-05-17 22:32:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950168-new-release","meta_rev_id":"446aa473c909c34ee452c4d41dd08b8d862b8945","_number":950168,"virtual_id_number":950168,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:44:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:29:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:44:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"22527c02322c19f791d81c7f6f8279fe77f16c88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:35:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"301a56bff99068e8213e649f50e8dab8498dfa2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:29:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ab6ee2e66c4391b9c37036e7ec2b19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0babe6c412cb4ee0a9e2d6cebe7620b3 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae3eab6403e487fbe1813f65306c5c6 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5188d99472614577ae3a64464f9a1f10 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99f826eff919435bb38eb4c828deac00 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04beaebb525646199968023926582417 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea5a518bd60d40f591e498e4ab11da25 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0692a1c88644287abd68a07d18ba1a8 : SUCCESS in 3m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fca3cbab6c0a48bfb06fb1851ee6df1e : SUCCESS in 6m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d1fac8538d8402fa0d13ee1371d9bbf : SUCCESS in 10m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7d9541cd86684cd1b5ff79fadbb2be17 : SUCCESS in 9m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2313bb00a5c44be9a7a2f44e0af3e6a5 : SUCCESS in 20m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a596943c31454716ada0c556a9c2cb01 : SUCCESS in 10m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2596de4fd23a4718a5a701881b52b14a : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b210ab0f63f6459e8be3dbc79cd0dc07 : SUCCESS in 6m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ef139f8dfd5f4a36b789fbb59100f03d : SUCCESS in 5m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/62394da62dac4ea78da026d0115e44ba : SUCCESS in 7m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fcbb69dfa523406ea4a33680f59207d0 : SUCCESS in 5m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/379b8a39cf2e411fbeac25d911d6ee0f : SUCCESS in 7m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0a0aec95983047ad9efcc8e458e6f2f2 : SUCCESS in 9m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0292ee4eb3aa47d7b239f1ee642d12ee : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2fa4b78f2c2d43e3904e8e39b286df20 : SUCCESS in 10m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7d97ce6f3b1b41c7afc8b6692c55f176 : SUCCESS in 9m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c6a8273f7cb3432c8f931e6ca02b0c16 : SUCCESS in 13m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f0972efef43f4afda305f86d41ef3ab6 : SUCCESS in 3m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/451cc9d29bc74ab6bd143d9039c27c3e : SUCCESS in 13m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4cde284940ca44a6a16e05b790f31c79 : SUCCESS in 3m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/03d3336399f14c66b8eaf125d0915dcf : SUCCESS in 4m 47s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/730b4f4e9a364244a940593564869774 : SUCCESS in 10m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/45bc359ba14e45a79c224f1d3bc60811 : SUCCESS in 18m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/46062215e50d4beaaf284d46b99ce19d : SUCCESS in 7m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f48e4503c1b64fad88d42d9435dac356 : SUCCESS in 17m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/081bac394bf740d294dcaf06247a13a9 : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ebb84b4910c248239b8d67ac8507dfb2 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47ba2e4a0f2647dda06e702d811b416e : SUCCESS in 25m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fe0ad8a6289e4161bb7fbea7aaf64ec2 : SUCCESS in 7m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fc3cd750a7b54e9c89676667f51763d5 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4291edc7d5524d939684250fd6309ffc : SUCCESS in 6m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0cd1688d99a74cc3ae2aa794f4b37a72 : SUCCESS in 15m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18c14e3436b14a68830e88118d22840c : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce60443f3d994952aeb6574942b463a8 : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"23e609c5ece6b7f2ea4023b5a725bb7388dbd7af","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:44:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7261acd833702f3258df0a7985d56235bfc2f920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:44:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be683b1d1c22fd6199ebbbff6270b799b12d3e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:32:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b318baff21e3437995c51923b992c1bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0baf565790b149789a7be8b5c884c181 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b871b54077154da4968820091295accf : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c501cdb10614fceab08ba90d4c5f29c : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf9bb55e971a48c599a366cef771ffde : SUCCESS in 2m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c538ec51081f4cb8a0b2f9eae42ec795 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ecfbf038cfb4e4f943a3cdce9ce0ed6 : SUCCESS in 2m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/08d0d56cd7ea4432956269bc51b67ad6 : SUCCESS in 21m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d464682ed4ff44a9a2ae896dfd6c11d3 : SUCCESS in 8m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c8fb67b8b7ca42dba77e6c155ca2b2ef : SUCCESS in 18m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6a02ea875158407e974c0e114e20be4c : SUCCESS in 19m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/37c7d2da3c134a6c8bd54fc8fd792898 : SUCCESS in 12m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6dd4809e74d14a209ca6529a2a14c519 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/49d6598c8e9f46688039146f8bcccd80 : SUCCESS in 2m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/363357ff9062436d8ca25b6ba9d51123 : SUCCESS in 6m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c44467d47e0743b1a1944788c2944c07 : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/592347daa17e4957b52a10076f6daf0a : SUCCESS in 8m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1510d364d1284201839e514ff76df36d : SUCCESS in 9m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e75b133c0efe4db98a9c8188a2fbc2d7 : SUCCESS in 15m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3de0dae380a44078b1189068a4ee7845 : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ba9b23c2547f48b9a6cbce9c4ae4f51a : SUCCESS in 10m 16s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d80170444ab84b97ad192e0984d702ce : SUCCESS in 19m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7f6b9f1fe32f4288a9533d56f62cce05 : SUCCESS in 9m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/025870b3aa07499aa40ce37bf2957739 : SUCCESS in 15m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/edbcfb28e9534f4c8b77b8f9974c538e : SUCCESS in 2m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2bc15006a33b4c57bab5989462b6e361 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7fe625121c46457a99389c9759ccaa87 : SUCCESS in 30m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6f6c14343b854dfda8949956b39a6432 : SUCCESS in 6m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/229081c14c3f4d989cd977ea86fb68e9 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a16e5fddfbf7463d86f7bb0f8330f463 : SUCCESS in 6m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8a79543e8c1a4723af4043d94d9a5914 : SUCCESS in 17m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9a0097d68a54f1ea655f34bdec4bcf2 : SUCCESS in 41m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8088346e1bc245bc8c7fae9f3a90df49 : SUCCESS in 51m 08s","accounts_in_message":[],"_revision_number":1},{"id":"446aa473c909c34ee452c4d41dd08b8d862b8945","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:32:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0a16d3ed292803f4177f243c845babe50e20d04","revisions":{"f0a16d3ed292803f4177f243c845babe50e20d04":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:35:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/950168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/950168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/950168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/950168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/950168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/950168/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:25:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:25:31.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 11.5.0","message":"update constraint for python-neutronclient to new release 11.5.0\n\nmeta: version: 11.5.0\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I80ce63caf9898e3ec9d8ba20a1196c27b1a8f552\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I63656996e058cd7c8ab4f49e1529df2ec0d8eba3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0a16d3ed292803f4177f243c845babe50e20d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0a16d3ed292803f4177f243c845babe50e20d04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
