)]}'
{"id":"openstack%2Frequirements~950171","triplet_id":"openstack%2Frequirements~master~I031f92eb056c011fc83e9957ec5e5b1a80d5e17f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I031f92eb056c011fc83e9957ec5e5b1a80d5e17f","subject":"update constraint for oslo.rootwrap to new release 7.6.0","status":"MERGED","created":"2025-05-16 13:39:27.000000000","updated":"2025-05-17 22:32:32.000000000","submitted":"2025-05-17 22:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950171-new-release","meta_rev_id":"3d111d8d14d7608f265e4cccfd58d23e1d116043","_number":950171,"virtual_id_number":950171,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:44:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:27:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:44:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1d38fbe760af7d2bf43f79f62c873eea317cebca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:39:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a0747d698a06b563a9d543309fc4ce3a3e88e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:27:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd719c86e6044aff9b6d21cdc3aba880\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3da707c87384a9684648e30e69b0777 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94182ef08d6d4b62b471e0f9b1f39ee8 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64ba03f184fb4f0eb133eb0949be4485 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6f7daec8eb94edbb0199db0ddbcbf02 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24a11ca6d254487eb45da0f8b6bd1009 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36a9d1cb531f4ae6a0a6e38e10971ffb : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecc4415c7e3f4814bd124331f0f48f9e : SUCCESS in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b593facec5e4e1e918c19a6c66a52ed : SUCCESS in 15m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4f547e406f8e4de194f1e593b48b25e3 : SUCCESS in 10m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f8506cb008154c159637b824a6d32e1a : SUCCESS in 23m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a3a1e235fe00408aae843fa7d1ec46e6 : SUCCESS in 17m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/688301be7b144fa7a72c4f0e301c98a1 : SUCCESS in 22m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9118d3f4758b43019ae3224cdde7bf79 : SUCCESS in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/25f740535fcc4481b6df98ebb814f758 : SUCCESS in 3m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/def99ba1ad524942830ab049ed3db206 : SUCCESS in 5m 12s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/66000ba1fd9748efb278382825981720 : SUCCESS in 11m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/91981214d6df4316b98f84a3bf1dfc4a : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6b73bca1e1ee41c486be78b827d098fb : SUCCESS in 5m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/59ba96a9d33f4daa852876c2ac746200 : SUCCESS in 8m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c97a208fce75498b9a124cf8eaf90e98 : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3eefca4b1eec49608c51f84ffd5e2587 : SUCCESS in 5m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bacf926d64b549b1b1d1f1872ac55626 : SUCCESS in 9m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/193e289b70e84720ab146c0cd9977009 : SUCCESS in 12m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/de23ec3047134983af5720e37053dbbe : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/972abcd02f1d47be8c1e1e6f468aff14 : SUCCESS in 9m 05s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/534b8ff598f648479a203db426f2e0c8 : SUCCESS in 5m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a249f29b828943f9aa40e6644ad4e9b5 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/74801e58700545fdb054736f5de29865 : SUCCESS in 10m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7d4ba0f1791c4b83a340a09924be3ac6 : SUCCESS in 21m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/672c6bbf9ed54b43ba4e653e0a6200d8 : SUCCESS in 7m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c4be18cc6f934c94b3caff16b3ed8594 : SUCCESS in 9m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6709c4d9cfec4d76bf7b75956f815bcc : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6d0d1f28c834826a4ceb02e03f5a11d : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86bc1b065ceb4201a749cf25f807c918 : SUCCESS in 27m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a5f19a9c388344bdb4b17ddbe50c7ea1 : SUCCESS in 5m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/84ecbc30330a4dd6a971936ea2ec4b6c : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6f85b808df14e58b6a0d5b02d5d07f4 : SUCCESS in 4m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/87303b9ecc5b47cabe80550db4211622 : SUCCESS in 16m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2dd585e366554cfb84fcdf606b0b0bb5 : SUCCESS in 37m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/440685f7d99442d097bf08dd79cbc2b8 : SUCCESS in 1h 43m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9227385518766027b79ead904decd4aeb9bab3e2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:44:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df24cc679bc44623cd96eab920251fcaf5e4e144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:45:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ebea4b28bde3829aa302fb1a009ca5efb015b73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:32:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8804e96780384d328ef64f89a591678a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e654fcb52bb24f86a17f80e03ae48463 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10fffb16683d471caa380ca5687d868c : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99c16036bade4f18b25bf4883f2a1320 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e87893717d34aa79e815239025966a8 : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9190188ddd3d4097ace280fe7d63da2e : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be18fca0c8484c8386cede841ea38382 : SUCCESS in 4m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3072443bc249433dabc54d3faeb1dab5 : SUCCESS in 17m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7c467e89b530455f95c7819d8c452726 : SUCCESS in 10m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f29b89d4fc704d628727c50af453365a : SUCCESS in 17m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4d9c375531df4782a228840f72b8ef5a : SUCCESS in 11m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0db98842d5b1473bb6c6f0055131330d : SUCCESS in 12m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a07aae9522454a9ca9af77c4dd9d45f7 : SUCCESS in 4m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c3f4f13a217d4f5281331854dbe37346 : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fff3b87c02574fda9ec811c508ca1d9c : SUCCESS in 6m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dfc157f652e44c148b87b3402a1d1cf8 : SUCCESS in 5m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a6e5ea252a24390aae5b18b700742e2 : SUCCESS in 8m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/31970ab7a8e54c93981a72cb362514ef : SUCCESS in 9m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1ba432c9f255451984fc8876c640c8dc : SUCCESS in 11m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/deec0fabf1cf4d8eaba7e8b6ae5c9187 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bab7ce4ce8364140a66df2bdfc0a63e3 : SUCCESS in 10m 45s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0ceb915c1cef4f60868edb519df7038a : SUCCESS in 18m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2c61865a755a4c6194527a457347d200 : SUCCESS in 9m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/40dbdf02ea7b461b8d6f78407ad581b7 : SUCCESS in 13m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8463fa53203b4aa8b86a93a634c96975 : SUCCESS in 7m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3df3bc10edaf49e4b70a84547b14d784 : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0321f9264b5b4f8190edec6c26670059 : SUCCESS in 27m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/af4a8be2b280456cbab5c5168444b527 : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/375328f6941f4895a8e48e20d46a678e : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0f6afc4a7c048ceab16b70445e57911 : SUCCESS in 5m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ecf8ac4ed50e42f0a0f5299453f0085f : SUCCESS in 14m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8805da94d62451f881d0713b960d01f : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86546a866c754d60800d015a5c8f3a1f : SUCCESS in 1h 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3d111d8d14d7608f265e4cccfd58d23e1d116043","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:32:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3e4329cd2023faabab625ea018b4e5795a2ae08","revisions":{"c3e4329cd2023faabab625ea018b4e5795a2ae08":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:39:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/950171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/950171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/950171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/950171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/950171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/950171/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:29:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:29:25.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.6.0","message":"update constraint for oslo.rootwrap to new release 7.6.0\n\nmeta: version: 7.6.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I48f53237f8531d3a1023e21d556776eb0b1711b8\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I031f92eb056c011fc83e9957ec5e5b1a80d5e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3e4329cd2023faabab625ea018b4e5795a2ae08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3e4329cd2023faabab625ea018b4e5795a2ae08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
