)]}'
{"id":"openstack%2Frequirements~950174","triplet_id":"openstack%2Frequirements~master~I4a347d96e78d31b5acb5bbf20a3fcf712ed88aca","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4a347d96e78d31b5acb5bbf20a3fcf712ed88aca","subject":"update constraint for oslo.versionedobjects to new release 3.7.0","status":"MERGED","created":"2025-05-16 13:40:48.000000000","updated":"2025-05-17 22:44:54.000000000","submitted":"2025-05-17 22:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950174-new-release","meta_rev_id":"c409bfa619cfac5d476a75aa20fc351cdeb0eb77","_number":950174,"virtual_id_number":950174,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:56:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"feceb8e388115011f5f5d3f5bc4fcfe448c72818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:40:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1905a60bb5ca4bc88620f3d5a7bc4c46c4cde788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:01:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75e24ceb7e594363b67eccbbab51206c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cf458b070a94d869d8120d64d126e33 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4175a232755b4252a6570b8bf7adb8ca : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae2b4d60a3d7448c8918fed2dd1af5a7 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5b880085446431890850a91ca2d42aa : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bab8826968343c6974290ce2e5ce4bf : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b31e180c9734dd5abde114cab83c025 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdfdd9e89284491ebb49cfe9719ea20f : SUCCESS in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/03397f3890744f2eac566b26c687383c : SUCCESS in 15m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1c20cb39b9384ac7879dd765b1a2444f : SUCCESS in 12m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/673be1b265ba41a1b5c6792a602adc4d : SUCCESS in 23m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2d80ee10beb748d688340dc79fc076b1 : SUCCESS in 18m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e76683d1123f4486bf41a0351f9ae078 : SUCCESS in 29m 13s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/aee3ef2077e44540972d404f733f0476 : SUCCESS in 4m 09s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8a2016f01fcd4d2baacc22e8bae1e34d : SUCCESS in 5m 44s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/65d8ed75974a483f8c909b1408a38160 : SUCCESS in 3m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0b8aa80a006d48cf8dba841c0d589464 : SUCCESS in 12m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/301ee20e8156462e98dae1b267c00528 : SUCCESS in 4m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d0f9be895149404b820b84b76d9414f8 : SUCCESS in 7m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6c3edf31639b48728d339476106da894 : SUCCESS in 7m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/be7803330719467d9bcc140fedd1790c : SUCCESS in 15m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88de8961fc774aa699eb8d740c1d6ac5 : SUCCESS in 5m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ec6418d41e534c6d88fa7f71de557c65 : SUCCESS in 7m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/68e4d686e51348e1811f5b4d426078a3 : SUCCESS in 11m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8e24652f0ca449fca05cec3439cac309 : SUCCESS in 1m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bea6fe27125049bc9d3f9e3e7eba7b3d : SUCCESS in 8m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a064e99ed928445e8c6bc30892bf9d4b : SUCCESS in 4m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/977a04a514324fb1b2999619afe003aa : SUCCESS in 6m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/62f83244beb24489a46538f17ca422f5 : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1defc29fa784489db2e9b070da98e331 : SUCCESS in 20m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fa0123adb76b4b9db6570e84c58a8a01 : SUCCESS in 7m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7451a737da88430885c7d2bec88825f5 : SUCCESS in 14m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9f6b83c924f947d2a16928af3cf56f9a : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a38d945b216471893a75239d4280715 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae5254ee3be04499b0191d3a3a30b893 : SUCCESS in 25m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f2ec9c55c7354ab2916350517ff5f052 : SUCCESS in 5m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fd94105a7b7c4bfb8f81dcd02281c261 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/208b5ea04fb24401bdea9f759c830d1e : SUCCESS in 8m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c7cd9766ae594dc6ba0e15d4ddd22054 : SUCCESS in 21m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcf727a3e4d1445ebf09fc287c1086ca : SUCCESS in 1h 00m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7aeb932f7c02419082ed94a8608286a5 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0a33c250a12ac9250773da6e12ea8a6df124c5b9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:56:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ac1184c25c79ef7a300b957be6eab9064500aa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:57:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8cb382bdc57f760b8e6075287b183777eac49f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53135430ea6b4a83a29027eb0ec76842\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b16a24523046729d913be051283d8d : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0361312bcd584bc39d73c86b00aa5598 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82a7971527c84130b9aa970f1297b8df : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/133e4f64549f48dd880dc583d1cc14f8 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e519005410848319c174f96418af9c3 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f738bb3c0794d25a5a21f4ece871bec : SUCCESS in 4m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a8e95674a2e4605a1f1f6efb21a88f7 : SUCCESS in 16m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1f4e441404564e0a8c546e0997835011 : SUCCESS in 9m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/545caff7ede94454889319b8c8ca76f6 : SUCCESS in 7m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8fdfc3c5f61c4f0089dd27e0cb5ef50f : SUCCESS in 19m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/90b81e86c86f41409c5d70c1edba43be : SUCCESS in 8m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/11b2b4fdf3384cdd9dac6371151173cf : SUCCESS in 6m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/863ca819ab784c6181365488cbcb21f4 : SUCCESS in 5m 27s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/912ff7455c094154b56097cee2d27b95 : SUCCESS in 8m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2dc8dc8e2bab410d8d2a9f484ca3fab4 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/722b4bf9638943a596c2fb3a14f1062a : SUCCESS in 8m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ad8966bfd619468c8b2a4963a9574796 : SUCCESS in 8m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/79e080dac73c4cc4855ac07cd668d8e4 : SUCCESS in 7m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/565f2505cb084a41ad957c8a43120b55 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6e63c5c6010f4d3d9ad155856eb2f787 : SUCCESS in 7m 28s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/54eae1a25fee4ddb8f381bb0833e0a1a : SUCCESS in 15m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a18431a62f424d60aa09c0cd5532bd50 : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/32f6249352e7419583ff4358f5216017 : SUCCESS in 12m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/078325028e5646baa32e1bf74664a99c : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9115dd64cfc4058b940da0d5d7db795 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e19fa03b2b2e4b18a07cff455a8275ee : SUCCESS in 23m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/86b0c46543b84f2b98aa558a79acf7ea : SUCCESS in 6m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3539defe25d348db961c60ed0c9f861d : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/615a81190d1540f59c368c422d91f147 : SUCCESS in 3m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1394795133274b0b9e2d4cf5c6568443 : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db3f7ee0444e43c0ad48cee2e0194bf0 : SUCCESS in 47m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9674bc3185d2450ebf1468d95eade32e : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c409bfa619cfac5d476a75aa20fc351cdeb0eb77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e9da5741454e5035fc5fec276a7f08e10b90c4b","revisions":{"2e9da5741454e5035fc5fec276a7f08e10b90c4b":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:40:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/950174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/950174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/950174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/950174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/950174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/950174/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:30:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:30:46.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.7.0","message":"update constraint for oslo.versionedobjects to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7bd81c8948aa0d407e99f5e73f6ec2592c3b8993\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4a347d96e78d31b5acb5bbf20a3fcf712ed88aca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e9da5741454e5035fc5fec276a7f08e10b90c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e9da5741454e5035fc5fec276a7f08e10b90c4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
