)]}'
{"id":"openstack%2Frequirements~950177","triplet_id":"openstack%2Frequirements~master~Ic146cb392418b03e1d3b1aec4ed4a51928ceb882","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic146cb392418b03e1d3b1aec4ed4a51928ceb882","subject":"update constraint for keystoneauth1 to new release 5.11.0","status":"MERGED","created":"2025-05-16 13:43:37.000000000","updated":"2025-05-17 22:47:04.000000000","submitted":"2025-05-17 22:44:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950177-new-release","meta_rev_id":"c6fe67dd892a2f8fbadcef9c7594271a551fb330","_number":950177,"virtual_id_number":950177,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:44:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:59:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 14:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:59:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a5f66dad7f9e122d2e8bfb89eed317c3ca6fccdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:43:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55dfd42096e537bc4d4175a97e81f8615e69deec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 14:49:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab09e6714bf488abff7ee0a3c19e437\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7476037f5cda4b6493e55ae86107d57b : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c1a7903d45343c096f90637a506df88 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cbf249a6035466ba15e834f05ae0a1e : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d84613b79b84491680b45c8bbcfe2f1d : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77d5383a1b4b4224b188c73b4cb4e476 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6000261619d84d839bd409b2557b0339 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d2f562434aa45a38afcea17fbee4d63 : SUCCESS in 3m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f9bfe5cf848c4cf39d78e1a04b197d61 : SUCCESS in 13m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/13a0b0e13b8d4122b64656540d5228b8 : SUCCESS in 8m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a64a739d2cc54e86b5d5bf833dd5fd82 : SUCCESS in 18m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/96d1631be8af4c738255c27ce790e5e6 : SUCCESS in 17m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c800e9f9ede54a1b806ca8faef3e4916 : SUCCESS in 13m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ccc49a99c3a6495287791caa1cb44dcf : SUCCESS in 6m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cd38a8cb2887479bb5d0c6d164945f12 : SUCCESS in 4m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1f2bdb052b1b4b10afc134bf88eb0f1d : SUCCESS in 3m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/21b3764a587e404b8f401d9f5f33ced8 : SUCCESS in 8m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/172169ab0654493bb05211091ba0b3b5 : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0f73a586ae964efa8fa0c0fa34a04af5 : SUCCESS in 4m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/34f017f68fc944ddb8601d47f9832e51 : SUCCESS in 7m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ebef25b143dc4e07bd3553740f3df9d8 : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed8e70d6a0fa4cd4a0ccbc398567a408 : SUCCESS in 6m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/badef1421c834ff88ba27d305c935e28 : SUCCESS in 9m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d265116d83544643b1813ca5feab6f4a : SUCCESS in 10m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/01a338beb169495ca7731f08f44ea080 : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/289f367f44fe497e9650c9aadbb339f5 : SUCCESS in 9m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/524372ae1aa24cc39a7756e9efa7990f : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/759cf9078bd843c18ee95fbe12e9bc1f : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e458e6a7d6624d94b7bbded288678dc9 : SUCCESS in 10m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f060b94ae771412bb414657614d5779b : SUCCESS in 21m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/49641e8e7d0a4d398fc5468fc08f031e : SUCCESS in 9m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9a8eaacd83494b5087b707bf5a5b4aa9 : SUCCESS in 14m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f7a2b0ee28594af7bfc12a5e9c186202 : SUCCESS in 3m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/da74064bd0494b67ba64a45ccdcc16db : SUCCESS in 2m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/456d62d672a647fab6829b9813cb4e27 : SUCCESS in 27m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/747a4e43f5d54773b3dae1970dc8039e : SUCCESS in 6m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6074ea58e3a1472abf9e876f16b501ac : SUCCESS in 3m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2c95e9b30d74b9ba1687a116236ebd2 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/687a4534dae44a109621059af7d88560 : SUCCESS in 17m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/982a2867193948049d6ad43a68044634 : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2aea45d260e4ab8bcb734bc3826b31f : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9912c2380e943bbf99673ca76d26ab6b18b9ebbd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:59:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7fbf005e0d51d015c60a40a7f29b1966ece849e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 21:00:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e964493760daa80dfee8248e42efc9250abad17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15fed3ff105f459bafd9f345e1d09cc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a08d5037009409b8834fbaf3495ee4b : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e773cea8de274e4db5662f12422dbac3 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f5b16664fe74176aa87bab1124339e7 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6fd3603c7cb4771bf66b5fcc7858580 : SUCCESS in 2m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0828f1dd1cac4d119996dcfb181fbafd : SUCCESS in 2m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48d34c8480b64aafb03f1aab5d2be28b : SUCCESS in 3m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d5c986e99eef4e1092ab84015c7df695 : SUCCESS in 17m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7886dd5a099248e9a97cc22eea9e94b1 : SUCCESS in 4m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/92da0ad2a10e4176a2be66e65562abc7 : SUCCESS in 19m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/511b38fe03fd41dcb58a517e664fedfc : SUCCESS in 19m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2426f7438bdd4cbe82d9e7cfe85dcd86 : SUCCESS in 11m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8590342874414598afba5c2a05e9f323 : SUCCESS in 6m 22s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5fccdb2cc60e41a392432071887e9834 : SUCCESS in 5m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f91c9ca9bade45b1a91ba4490757ae22 : SUCCESS in 7m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7b53dc49ccc3488f8a5852e4a8281402 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be70f21a2ff445b19a59bbf55909bc51 : SUCCESS in 6m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d906dd4efdc541d6ab31a1adc707e570 : SUCCESS in 5m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a1543cabbed94b548d897366af8dbbfa : SUCCESS in 12m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/94ce84c958e8455fac68922fb85fa3be : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0563212e713c4456ab09148369069dbd : SUCCESS in 10m 54s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/53f365274aa44cf2b5d0b69df8b80b1b : SUCCESS in 19m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1a920480272b46f5ad80bc15d660ae0c : SUCCESS in 4m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/59f3e901b24344d6ad65456b55a0acd5 : SUCCESS in 14m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e314dee1c5334c21b54d75fde4a48607 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa99019c4126482d9bacd903eb95389c : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6d54672373b45ca801cc52914aa5b4d : SUCCESS in 21m 31s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/db73f7a3bb8642dcb861ddeaf8bdae55 : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3c8efd28728e497ab987ac9e7d72e6c8 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad2be72cbb0e438bb50331a02d0d1b4f : SUCCESS in 4m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fe42a934bd1842db90ef83a3c083d5ac : SUCCESS in 20m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/592f10d7a6b143dab9b3db0d196b17f8 : SUCCESS in 49m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6935d46d19f5440faf3736a538e44569 : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6ce4846cad925f25950d2f7ed3ddf49f33c5d842","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:44:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6fe67dd892a2f8fbadcef9c7594271a551fb330","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:47:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7c47e4eae6947efa2b38cf96d10e7ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8fdece5417459ca414eb0e41464886 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0be7f72c1693102c164bc8e15412ca78d4b2a17","revisions":{"b0be7f72c1693102c164bc8e15412ca78d4b2a17":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:43:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/950177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/950177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/950177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/950177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/950177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/950177/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:33:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:33:36.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 5.11.0","message":"update constraint for keystoneauth1 to new release 5.11.0\n\nmeta: version: 5.11.0\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I4915728ed5a7dc1ac1625fcdd45d44f3fba5b803\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic146cb392418b03e1d3b1aec4ed4a51928ceb882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0be7f72c1693102c164bc8e15412ca78d4b2a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0be7f72c1693102c164bc8e15412ca78d4b2a17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
