)]}'
{"id":"openstack%2Frequirements~950179","triplet_id":"openstack%2Frequirements~master~I162bfe29cf538b72ddb01ce7bf3404278374af42","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I162bfe29cf538b72ddb01ce7bf3404278374af42","subject":"update constraint for neutron-lib to new release 3.19.0","status":"MERGED","created":"2025-05-16 13:44:32.000000000","updated":"2025-05-17 22:33:40.000000000","submitted":"2025-05-17 22:32:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950179-new-release","meta_rev_id":"d78165caf04cdc7670eaf27783e6cde94d6159b8","_number":950179,"virtual_id_number":950179,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:44:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:44:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"031adca71ff4b67f8abdea95a1631a0b5824fdb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 13:44:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46061a6a78ba31e1a5eb5c022927016e8dfa7c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:26:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e108027f58447c3872a2ff758760369\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac544db524d4c72ace08064401cf11d : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445c06e4580040198d05e911a8485f9c : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fe37a6da85648c2b177ebf725501cbc : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0bad51b30b04a0bad919536bf2aaafd : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a53fcd9bdefd46bfab6c7e10e5153dfb : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4985802ac7304f099860edceed30e902 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/625ca1c9f8df42d3b6dfb5390c818a3b : SUCCESS in 3m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/55996972f75d467ba954cee007e4f932 : SUCCESS in 19m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d910138afe7043038ab20a024bc8d4a6 : SUCCESS in 10m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3fff9fc83c24457098b17d769ce74469 : SUCCESS in 16m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/45d652e0b63041eb8dba560e4b93bc74 : SUCCESS in 25m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e66872a39bfa40d587e9f8290ab2b4a9 : SUCCESS in 10m 30s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/03f115a510b74ed8a6b3eca5b8fce12f : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2b367b4505434fe49aea86d6cf3262a5 : SUCCESS in 6m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a4f738450bcf4f629b37f675d3f9b38a : SUCCESS in 4m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/241076b24cdf4172870ac6a72eb026e3 : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/73391522b1a545928fd4c5e3a65c8d8a : SUCCESS in 6m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f650f9074b7d47d3866d9ac357446bf8 : SUCCESS in 5m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/228b573d3cc54cd59f88e922e6542524 : SUCCESS in 9m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/50e3a784357048ad8266abaffb297e29 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7e743f43d4a40a09a25934d9db4deda : SUCCESS in 6m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2db07d8fa91c4c9d9cb262d828d58d6f : SUCCESS in 6m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3cc77838ec9e4fca927d6e2ad72aa743 : SUCCESS in 12m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1b23e783cadc4ccea94689dd6e6ab8b6 : SUCCESS in 3m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2a0dbfc303684cef95ae8dd906dfe83a : SUCCESS in 7m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e7c2cfac5d83453bb09fd1a8997860cf : SUCCESS in 5m 55s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9440421322b94480b6a70dd308ec3454 : SUCCESS in 5m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b4a66e10cea0452d9c561f429d66205b : SUCCESS in 11m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3be8651885dc4136bae92c27cce56f80 : SUCCESS in 22m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9039f1cadb084beabcada17aba5931b5 : SUCCESS in 4m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/849f6679011a401692ddbb18c40bdeb5 : SUCCESS in 14m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a795797192fc41899f035d81b14c6842 : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57e26ab92c9f432997001edad5bd7d5e : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c58026b782c45cea9debe7169e534d5 : SUCCESS in 25m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/eded36ef37fc41eeb6c23f0b6c5af0d3 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e7654d2302dc41d08650de710f09c6bf : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83be1298b1aa44fb8b3d0d40d26b8154 : SUCCESS in 6m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/623b56254e944cf1b1bc76c1e7d07384 : SUCCESS in 20m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/604dcd53999745f89d103daf3ae5eba1 : SUCCESS in 58m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b813a7ca5f14beab07d02d0b54bd2c0 : SUCCESS in 1h 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fd083811af1f9f9698a94477f874889a5ce837fe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:44:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"072a635c8b96f2e6575078f84393591b7e46ed49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:45:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7738b9af37bcc8e6203f60eb9a6fcae8ba47d333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:32:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165d2ef47cbb45f5afdcf5b3ed900a94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72072b5c15334d3aacc48ad48d46b1d3 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f0d119b187469ba3e151eb7a5a2dbd : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eab1c8be48aa47528c3900f43927b5e2 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4c576daf2804d2d90697218d539920f : SUCCESS in 2m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0295dd8f87cf4f35b9f0aeb451798c39 : SUCCESS in 2m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a49f94fb09e42a8927148b918e53c75 : SUCCESS in 4m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/04d319d30e854e6ab29719e9ba655472 : SUCCESS in 6m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eb41d980064145f4a15ee217a02ab584 : SUCCESS in 9m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/930b123a550e4216b2a593f456338510 : SUCCESS in 27m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3ecd9e5bc49a4950a9ee22609a8a04c7 : SUCCESS in 20m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4c13f64182d74c6abe411f0b75c04911 : SUCCESS in 11m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4294e18f47d74c798d5765281483c12a : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8c75ace467fc4abc8c4adcd80e8afa5f : SUCCESS in 5m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/21af783609b64d869cafa01ec7b44bbc : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/de9e214ad8b045ccbf17029f5ed37619 : SUCCESS in 14m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2f8dafd94894aaa988be5dce1cef56e : SUCCESS in 14m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/241eaba9b940498fa33183f940450492 : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f06cddbe9273495c9b2de97c213b1984 : SUCCESS in 12m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f621a5eb03744bf6818e3e93148972c3 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e2dde5b5be474a6ca53a0fee148d8764 : SUCCESS in 9m 06s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/70574218e6e34dc1ae4dbd34d931a5da : SUCCESS in 21m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ea287d31fed44721b528968e33692555 : SUCCESS in 7m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b2d7872be375443eba7825cdc63c54e6 : SUCCESS in 11m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8f635d213ad14d469e97675f7e9c6320 : SUCCESS in 7m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca848c4d745f42d9b1c758df6f73e0b4 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a4908a8d1244492a6d01720c321fce4 : SUCCESS in 26m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3ea745a8ce6f40b8a5c2a5d63bf68997 : SUCCESS in 3m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/58016d46e0644a9db20107ad922b705a : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/291a717358744f6d9cab84bf35a282e0 : SUCCESS in 4m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1ecc6b8f7a0a4056a1248572dbdf7b38 : SUCCESS in 24m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b494b27809b641789a1289c0f50f7aa6 : SUCCESS in 1h 19m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/228c8a1570b446fdbfcb7ee6d60d0c94 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"61098bc85b63e53896b00b39c5611e8ef27e59d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:32:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d78165caf04cdc7670eaf27783e6cde94d6159b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:33:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d9e3a1d3b194de39504f65e8b27f313\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59218008da4842c6b142de40e479d628 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5163ceb6aaa567cd5d1ed903963b50899308707","revisions":{"a5163ceb6aaa567cd5d1ed903963b50899308707":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:44:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/950179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/950179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/950179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/950179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/950179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/950179/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:34:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:34:29.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.19.0","message":"update constraint for neutron-lib to new release 3.19.0\n\nmeta: version: 3.19.0\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id6cfa5b563f58e3335b6e5c4878fc64897537e35\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nChange-Id: I162bfe29cf538b72ddb01ce7bf3404278374af42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5163ceb6aaa567cd5d1ed903963b50899308707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5163ceb6aaa567cd5d1ed903963b50899308707"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
