)]}'
{"id":"openstack%2Frequirements~950185","triplet_id":"openstack%2Frequirements~master~Ie870cbc0e21dc91bd0cd120027ca5bcac09720a0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie870cbc0e21dc91bd0cd120027ca5bcac09720a0","subject":"update constraint for oslo.messaging to new release 17.0.0","status":"MERGED","created":"2025-05-16 14:07:37.000000000","updated":"2025-05-17 22:16:14.000000000","submitted":"2025-05-17 22:15:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950185-new-release","meta_rev_id":"bded3693a8612f2a0593f53dce39d79e008c68bb","_number":950185,"virtual_id_number":950185,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 22:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-17 20:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-17 20:42:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 15:51:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-17 20:42:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"dcf5d59aa83c3cd8166667f1a70b262b6074c55f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-16 14:07:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5de6c9addca04404d7de7fc726ed5e1fd105d90d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:51:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84aeae4890284cbd9c28aae23d99f75f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0fb3b220ba490ebb50ae5e416ff3a4 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975dd42f20d642f88ad3822039e2a7bd : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c3573ef9927407a8d2edd53768ce545 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b22bbf1f38784d74b6b0ca75bc4eea39 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae28486fbf0a4bfaa2deb37d114f8f2c : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe0175e1629349fe9e6b75ed69a3f54c : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa4c3e2aa1ae4dbf9c2f94fb0602826e : SUCCESS in 3m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/67836ca964b446cfa455beaa66f15108 : SUCCESS in 18m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3c6a456736de4f9aafe86f0a8e86820c : SUCCESS in 4m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dd182c21236f48809f053843bfe7c4f3 : SUCCESS in 17m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/358338be9c554143a6263d11cfa67027 : SUCCESS in 8m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/55f21d65e7eb447e8d738991fe391f43 : SUCCESS in 23m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8263e580e8664ff3a508ffbf46faf06a : SUCCESS in 3m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f8b43ac01f354af698ceeed0729c4264 : SUCCESS in 5m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e02bf12ed47b42ec8cdf020396ef11f6 : SUCCESS in 3m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/41372e7e5a7f4677b88580403ac47401 : SUCCESS in 9m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dada9685d02b4be1a3e97ebd9370837a : SUCCESS in 4m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6fd9747b8aaa4eb08b9a58a83a30d911 : SUCCESS in 4m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b8c06975e85b49739f3463e376003162 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b250000e890c43a899838631ad0fa2e5 : SUCCESS in 13m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34ee30f4a0c740e08d8ff70c0d9f0c85 : SUCCESS in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f36e676ff50642359de5eaf00828a983 : SUCCESS in 7m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d1cec31605ec4fcd9c6fc3cfe284bd7e : SUCCESS in 13m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7a1a73a938a345588aa1ab549f62d5e6 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d5bd053deb86407f91b70067bcc0cd48 : SUCCESS in 11m 33s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b5937f512fa4480f9cd1c1a9b685043f : SUCCESS in 6m 13s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3e9f1b9805d04acaa335a8092e4a2356 : SUCCESS in 6m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6eb5d225eafc4867a9f8ae185d2459e0 : SUCCESS in 12m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cd10c0cc4a0948cdb81e515ba67752c0 : SUCCESS in 15m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d601d326dd664fcfa8df9ba6335e57aa : SUCCESS in 8m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/65b6c9c12ffa4129a9f91e1f0a7eca33 : SUCCESS in 12m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9a00cb3d5bef42388e1e1adc2f494eb5 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/152c1885b1014df2a23aa24e5caa5d9b : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0144eb0b3fdb4209a7cfef840cd127a9 : SUCCESS in 27m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d17e2b500896471e8a47b106de7b29c5 : SUCCESS in 7m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8572975565d34607a4bf4f4744a31c53 : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a2c07e362a643bab3ca3ed25d3fe1f8 : SUCCESS in 5m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5aa721e24ea44e188e8aa1e4936a08c8 : SUCCESS in 19m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26139ba4387d4e8db098d13a6004a715 : SUCCESS in 33m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/793e76f2a4ec4c47af57d3f88952d876 : SUCCESS in 1h 33m 55s","accounts_in_message":[],"_revision_number":1},{"id":"96510ff895724abf0ba80ecb95132c47dfe77acd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-17 20:42:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fab77a9bb44d807202e6dc57dd06f46a3857d006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 20:43:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bf2d0e774d9be9ac6784b13bafccc5c4c9d8e73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:15:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b3d11ccb82341c99ed493e3371fe685\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68702d019f62439dae0049fdfd1cef43 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3333127e73f04e4fbbade97bef17d755 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12a15f9369cd4dc597b33563c5d0578e : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb087321b79b4ac6bf12fdad2f49b5f5 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24385110bb364fdf9cab16e9d2f9720d : SUCCESS in 1m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4c2ca3a7a0040f6b283d46805f9f38f : SUCCESS in 3m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f374cdc394ea43e5be19e2f768aee80d : SUCCESS in 6m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8de597ddc3994f40b844b007cc876668 : SUCCESS in 8m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e37451e28baf4c9f831be2aae9551c9e : SUCCESS in 21m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/87d7edf5f0cc4c0ebd301350e9ed0d98 : SUCCESS in 22m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/729c9491780d45a2a05c8484e0ec6295 : SUCCESS in 10m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/251fb4ec154146f491c5daefd891cd9f : SUCCESS in 4m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1374e52b83bb45acb465330456e8ef2a : SUCCESS in 2m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/87e58a7c57844817a7f91b25b5e32027 : SUCCESS in 7m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b8fcc639d8334f8fadf40741f0ac61e3 : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08954218f40249fe83fb34e1834195b9 : SUCCESS in 4m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4878a9f48b6b4172899b356d228a406f : SUCCESS in 7m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/79b6f72140574e6e94fd9ff094d019ad : SUCCESS in 7m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b2a6fbe8d90446e6b7fbae7de0ac84a6 : SUCCESS in 1m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/774fb7ec4c9c440e9aa4b7236c9509ab : SUCCESS in 10m 43s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7adad221719b44128c699518fa573c44 : SUCCESS in 20m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b400e61cd7bb4c5d8c26d2b7b96e1ec0 : SUCCESS in 9m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/03cdb335c0754d24baadbee8e668d725 : SUCCESS in 13m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4b7889a1736c4baa83da4b42d905a9aa : SUCCESS in 3m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ee3d3f33fcc452c8380252edb3657eb : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a150cd7eeaf4fc999766750a470f7a3 : SUCCESS in 24m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3886b31a6a184db49e4a2719d7fcc0a1 : SUCCESS in 3m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2792a9eb40fb4540b77ee897110ad760 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1caa2d71e7264131b503a00d1dc279cd : SUCCESS in 5m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7399f230f30d415f8a119bd2af1eb073 : SUCCESS in 18m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/629a67c6576b4f7c9fbe12d81c8e1557 : SUCCESS in 56m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3361183a45e64b0d93643875af7c19fc : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b91be5743c3b042f916ebabefe6f4f4d3fb7fccb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:15:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bded3693a8612f2a0593f53dce39d79e008c68bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 22:16:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b63608ab1a940c594037fd051702def\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8b96db68fc4cf59480f8495c3710bb : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85ae3ace54fc78d09efb28429665d6f2569e7adc","revisions":{"85ae3ace54fc78d09efb28429665d6f2569e7adc":{"kind":"REWORK","_number":1,"created":"2025-05-16 14:07:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/950185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/950185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/950185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/950185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/950185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/950185/1"}}},"commit":{"parents":[{"commit":"eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9","subject":"Merge \"update constraint for aodhclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eebb7f4ef7201abc5a5407e8ad1ab51d904ea8c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:57:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-16 13:57:36.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 17.0.0","message":"update constraint for oslo.messaging to new release 17.0.0\n\nmeta: version: 17.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I6ec212552ac05dbe55be9634a7978c473e9317cb\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie870cbc0e21dc91bd0cd120027ca5bcac09720a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85ae3ace54fc78d09efb28429665d6f2569e7adc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85ae3ace54fc78d09efb28429665d6f2569e7adc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
