)]}'
{"id":"openstack%2Frequirements~950400","triplet_id":"openstack%2Frequirements~master~I95204af77570ba33ce13958aa3b8a8e49cdcf6d5","project":"openstack/requirements","branch":"master","topic":"update-font-awesome","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-26 16:14:58.000000000","reason":"Change was abandoned"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-26 16:14:58.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-21 14:34:35.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-02-26 16:14:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I95204af77570ba33ce13958aa3b8a8e49cdcf6d5","subject":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1","status":"ABANDONED","created":"2025-05-20 11:06:35.000000000","updated":"2026-02-26 16:14:58.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5bc94c81076257788a9ab250008ee08237cea25c","_number":950400,"virtual_id_number":950400,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-20 17:01:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-21 14:04:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-20 12:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 13:16:49.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-05-20 13:23:13.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-21 14:04:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-21 14:04:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-22 21:15:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"2b1297204accbdef7e93ddb0355adeb4f41f6563","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-20 11:06:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99f537e6cdcab707d011c3bd7943c0f7400021a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-20 11:10:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2a409684fa8e2eb5a5d32a2b42a4f781bb47d35f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-20 11:15:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"f226a17406ae912de1c0173d628610a84d965b30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-20 12:03:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6f6755a40b60c8d8ca56df92db4e30c83ac44ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 12:55:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3ec20fcb73a42a9bd8dda549871ae53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d75fb76fbe4c4da05a7dcc4dc5f0c7 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8702599b516419aa07add22cc963466 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ddf86b8b3234c5aa9d3a1e1b7d87409 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/767e574895ab4338a221d66beb2ab593 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5503c5956694c738d2987440d515413 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/634acad844574018a2683e8dc0f00b95 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8c73610b82443d2a582ba3101934f49 : SUCCESS in 2m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9b7acbb7786743d189e9ef77fd6ad1a9 : SUCCESS in 16m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/42b5bcaf39aa48f5ae9fdd378e860b46 : SUCCESS in 8m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/60e1a27a603d4ba99a87653b4017fe0a : SUCCESS in 17m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2a4271ffef954e87abf2135ac858fa9a : SUCCESS in 19m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/961db784d7714dd4ba20369c4b3ad01b : SUCCESS in 23m 51s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d0882ceada294735a73195373e9ac7d8 : SUCCESS in 5m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/11d458ad1a9f4ffca6b48b74de585ef8 : SUCCESS in 6m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bf6a058cc95548e99c7d7b4e54414951 : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7d5c724541734070929bcff5630c4ed5 : SUCCESS in 11m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bbb218218f414908895d7f3cfae7c58c : SUCCESS in 6m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fd34d6a5ae36408ba3ad537bd7aefb31 : SUCCESS in 7m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/62d0d7d0f3c44d08b60966e962306fd5 : SUCCESS in 7m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6ccee31e84954dbb8b55e0e85389d4e9 : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a706884cecce47f59d118a058c84d6b9 : SUCCESS in 8m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6478feaf2c87450997112c62a8f242fd : SUCCESS in 7m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/da5a7b81619b46eb853bed6d5c75acfb : SUCCESS in 13m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2ebebfe4756e47ae90042ea90f484224 : SUCCESS in 3m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/529a8c6729f04809a7b3832b872710e6 : SUCCESS in 6m 05s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/03762ecedc0b462f95318d0cbed613d1 : SUCCESS in 4m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c0114b0b0faf44acb2d8a8c7866773f8 : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/84a76f8d40294378a21f747c5109977e : SUCCESS in 9m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/54ae6bc822d74d989497f604091b44b8 : SUCCESS in 20m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/da738394fc434838ab1a8431e18bec2a : SUCCESS in 3m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/85b7370453334a0189452644d5027b44 : SUCCESS in 14m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/442f17ffe3ed46d9891d2930f91796f1 : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e56425e36b08441daf84bdf6e92433f2 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d387c51e8bb45208668fbf110fe4fa3 : SUCCESS in 24m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/129e8dca09ce41b8af33708c03b41f1c : SUCCESS in 6m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a68ec8e7256d497da79ff689151e9fc3 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/498bf8d72566424cb351e75a5a540f42 : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9893034645ac421c995d791c863f8c81 : SUCCESS in 17m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/492df04f2dad4952b07d3dc38b358bfd : SUCCESS in 49m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c015c1bfa9142dbadbce19333c88ba6 : FAILURE in 27m 43s","accounts_in_message":[],"_revision_number":3},{"id":"008352f23613a743c20299a32cc31bb4f424fbf7","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-20 13:16:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"551fe51b37c77c708568bf357ae9ac2f886f74fb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-20 13:20:23.000000000","message":"Topic set to update-font-awesome","accounts_in_message":[],"_revision_number":3},{"id":"c079c35c137e0731ba568a115cb1cf35f5a5216e","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-20 13:23:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1a11edeadc124a64c061709a1fa06e12874dffc3","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-20 13:59:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0349582aa1052c3be2e24a815d57030884afefc2","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-20 14:02:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c29ea7cc5216d1d7b418f0f14d51695a89724db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-20 14:13:32.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3931a2dc200e642276e8a4410a4a41eed773144b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 14:13:34.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e534ab878a0542fea5ae8326fe0f16c8\n\n\nWarning:\n  Dependency cycle detected and project openstack/horizon doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"452c6f6edc3a421f585a21009bbe356aa7a88407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 14:13:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/205470e7437745eb994eb27e408f7d50\n\n\nWarning:\n  Dependency cycle detected and project openstack/horizon doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"3ba87003eb5be7612fb8d84c275c6055f7719eec","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-20 14:26:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f6b68ce392aa0f5959e3e68ed70c0fe79618a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 17:01:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e63e647c4df434abf40a365771892ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cedd6e1f40a423fabcbab25e09d296c : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c17687e43cf84e1895f6971128ec8a14 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a87a81551244ddbb4c43871f81b4e9d : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d0781410cca41039d4e6e24bc1321f0 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8cf82d8eede4c9981251a9ba6cdf0b1 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1858abce789c412d8a1dd9ecd308bf15 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0d22a8ca682c46758dd1fc97f9cef9cd : SUCCESS in 3m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cb607d68fd3d4a6799a0b695ab59f6c9 : SUCCESS in 15m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/573a12e3732844659b4bfe2cd4cc3890 : SUCCESS in 10m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4bb3fdb358884c498789386edf20b323 : SUCCESS in 24m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/494347b415a94337bc5c0286c5468d07 : SUCCESS in 20m 29s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/49f8208dd6de4027bb5d754793082be3 : SUCCESS in 23m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c985511cd3764281aa387baf659019b6 : SUCCESS in 6m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/47bdefb3222b4176a03e040eb1de0fe3 : SUCCESS in 4m 32s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/65704347d54947edae395ba06f494cad : SUCCESS in 6m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4d7fbac7458847e58971be3f1b614ad7 : SUCCESS in 11m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ecd4693c67bc41e5b4e83dff97e84163 : SUCCESS in 3m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1883bc6ca659447d8f4f522497044ffc : SUCCESS in 5m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0649726caf0c4310bce7a2fc368034b7 : SUCCESS in 7m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/40e07e22304c4332ab391b422fd52a99 : SUCCESS in 14m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fcdb7ef6d6c54b2d83c4ef41fbbe3213 : SUCCESS in 5m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7bc03fff119741bb8ad10944c2549bcc : SUCCESS in 8m 25s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/957cc5fcc1ac45e9b4e4ce5ba56078b5 : SUCCESS in 12m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/781a2595b145401591a12367c4a0b44e : SUCCESS in 2m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cf355acbd73b4b70b12f8b14a131cef9 : SUCCESS in 10m 38s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c1aed596c65642cb9c096ef81c700480 : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2d923bad446e462a8bf3a312ef280f8f : SUCCESS in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bdfa3e593dbb47468d3ba543c151be57 : SUCCESS in 11m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a1d761add2b04614bde11f82a99ee9e2 : SUCCESS in 18m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/93c694ab5f504fadae6448a889b4e980 : SUCCESS in 8m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a686d0a498c240118fbfe52e0674f3a0 : SUCCESS in 20m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/716f09bd017c4d0da54e37be801c50d5 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc546be13aba4bb5b64fef9baa4b445a : SUCCESS in 2m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b137c592f6994529988e48d62d860c7c : SUCCESS in 26m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/63f430e7b2424c1e8c730735d4cee5d0 : SUCCESS in 6m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/de41e14a67e94c4e956cbe583083ec26 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db485e70ac774f9cb032213890aa9ed5 : SUCCESS in 6m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6bd9042ac63543e4bfa9e5f218825d66 : SUCCESS in 18m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a46951235c714c928647fd33d0009a24 : SUCCESS in 58m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4f99cc9140d4c70b329f000e19f2811 : SUCCESS in 1h 44m 12s","accounts_in_message":[],"_revision_number":4},{"id":"ab92feac4eae9b3a71b80f499dc051f67845669d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-21 05:15:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6671e39f2a2de8feaaef0651db97b7ed1b5be8f","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-21 11:34:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12e1bc7c71e9f535aeff7a824d50ae484cebf415","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-21 14:04:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"653e0675693a2c6adb988a5d8ee74e5c6965685a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 07:50:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5bc94c81076257788a9ab250008ee08237cea25c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-26 16:14:58.000000000","message":"Abandoned\n\nsorted in another change","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"05a34fac2ed827d53d9172fc9a6baa098a566b9b","revisions":{"5e0103c1e88ed5abdac111b4bd5aeef00a85672c":{"kind":"REWORK","_number":1,"created":"2025-05-20 11:06:35.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/00/950400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/950400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/950400/1"}}},"commit":{"parents":[{"commit":"e5a53204d07d849e58e693b6bfbe89a695a2fc0c","subject":"Merge \"update constraint for oslo.context to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5a53204d07d849e58e693b6bfbe89a695a2fc0c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-20 11:06:35.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-20 11:06:35.000000000","tz":0},"subject":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1","message":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1\n\nChange-Id: I95204af77570ba33ce13958aa3b8a8e49cdcf6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e0103c1e88ed5abdac111b4bd5aeef00a85672c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e0103c1e88ed5abdac111b4bd5aeef00a85672c"}]},"branch":"refs/heads/master"},"8a2353527943f7ac33c503d8c048f99772a21e50":{"kind":"REWORK","_number":2,"created":"2025-05-20 11:10:49.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/00/950400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/950400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/950400/2"}}},"commit":{"parents":[{"commit":"e5a53204d07d849e58e693b6bfbe89a695a2fc0c","subject":"Merge \"update constraint for oslo.context to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5a53204d07d849e58e693b6bfbe89a695a2fc0c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-20 11:06:35.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-20 11:08:56.000000000","tz":0},"subject":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1","message":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1\n\nChange-Id: I95204af77570ba33ce13958aa3b8a8e49cdcf6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a2353527943f7ac33c503d8c048f99772a21e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a2353527943f7ac33c503d8c048f99772a21e50"}]},"branch":"refs/heads/master"},"04cd1d0b561dc14f84b67c24842c88e7452ba1ce":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-20 12:03:01.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/00/950400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/950400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/950400/3"}}},"commit":{"parents":[{"commit":"e5a53204d07d849e58e693b6bfbe89a695a2fc0c","subject":"Merge \"update constraint for oslo.context to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5a53204d07d849e58e693b6bfbe89a695a2fc0c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-20 11:06:35.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-20 12:03:01.000000000","tz":0},"subject":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1","message":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1\n\nRaising the upper constraint for XStatic-Font-Awesome to allow compatibility with version 6, enabling the use of newer icon features and ensuring future compatibility.\n\nChange-Id: I95204af77570ba33ce13958aa3b8a8e49cdcf6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04cd1d0b561dc14f84b67c24842c88e7452ba1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04cd1d0b561dc14f84b67c24842c88e7452ba1ce"}]},"branch":"refs/heads/master","description":"Edit commit message"},"05a34fac2ed827d53d9172fc9a6baa098a566b9b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-05-20 14:13:32.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/00/950400/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/950400/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/950400/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/950400/4"}}},"commit":{"parents":[{"commit":"e5a53204d07d849e58e693b6bfbe89a695a2fc0c","subject":"Merge \"update constraint for oslo.context to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5a53204d07d849e58e693b6bfbe89a695a2fc0c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-20 11:06:35.000000000","tz":0},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2025-05-20 14:13:32.000000000","tz":0},"subject":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1","message":"update constraint for XStatic-Font-Awesome to new release 6.2.1.1\n\nRaising the upper constraint for XStatic-Font-Awesome to allow compatibility with version 6, enabling the use of newer icon features and ensuring future compatibility.\n\nChange-Id: I95204af77570ba33ce13958aa3b8a8e49cdcf6d5\nDepends-On: I911a1d6f3bcd851a5938dd2791171dd9b40351c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05a34fac2ed827d53d9172fc9a6baa098a566b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05a34fac2ed827d53d9172fc9a6baa098a566b9b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
