)]}'
{"id":"openstack%2Frequirements~950626","triplet_id":"openstack%2Frequirements~master~I138ee2ddd187be2100579f6a055a0a7d3c7aa973","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I138ee2ddd187be2100579f6a055a0a7d3c7aa973","subject":"update constraint for python-swiftclient to new release 4.8.0","status":"MERGED","created":"2025-05-22 07:32:42.000000000","updated":"2025-05-22 11:05:47.000000000","submitted":"2025-05-22 11:04:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950626-new-release","meta_rev_id":"aa5ed8adbd86aeb0487663acf33966c74119478b","_number":950626,"virtual_id_number":950626,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 11:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-22 07:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-22 07:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-22 07:58:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-22 09:12:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffe20dcc4778a8a5f1775f5ec4ee13d5f8e81679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-22 07:32:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1108ab37c79d12d236b964db58dd70e5932a27b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-22 07:58:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5881c2ca32a3abb57ce6721c026c1e460c9adb64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:12:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50b34ddde3594dda82b5515aee62d0eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483fa94b42aa461890f9ef12a9b25990 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85e60a07578640bbb2f86a5e9d8ef107 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a763ec62db649f4b3f90f0522c533be : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc8cf738560c425a8423b397446925da : SUCCESS in 1m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f41f5da7f9a541db828b264800518396 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0f0a89fe38f4c5d9b267bfe972d3e98 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/759b6bc377c34d52ae9ca4be969358f6 : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/30e8c8f53157417e9b16b8c62f9440c0 : SUCCESS in 16m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/257020373cc84d5fadd6526ee04fdf60 : SUCCESS in 11m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8b9dff7fb7b14c03a795af52e5a0311b : SUCCESS in 21m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e060e6c09ecf4fa88a3af2ebe1de552a : SUCCESS in 16m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2aedf26a7c1d47048ae071ee23b0c3aa : SUCCESS in 23m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e768eb9825b745268ec496b6ef237e07 : SUCCESS in 4m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f878b092224b4f759a7f2ee81652c579 : SUCCESS in 4m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d29b565ce0004270813d66a73ed4041f : SUCCESS in 3m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/eaddaba8b9014ea4ba97fe6edbb4f7ac : SUCCESS in 9m 58s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4d85ab90ff494b18ab5587dfee512320 : SUCCESS in 4m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/179c61f58d844d95893f425336a06512 : SUCCESS in 6m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/de4e3ab7fa264385ab962903d9be4d03 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ee8173ae03ea4161819e7a84c910844a : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd058771e5d248bbaf7e80e2e2048bc1 : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6ecc8b84672a43e88a7bfa83d55a7fe4 : SUCCESS in 9m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/36f8d4543ea2448b9db3bcece52abb9e : SUCCESS in 12m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/18cc055f749c4beca63aa2568ec14110 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9995deb63a6249038a03a7a036c8d77f : SUCCESS in 9m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bf4601b859ab4bb4939fb0220c9f1ea6 : SUCCESS in 4m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/89b1692a7c384fa0ab4cde0b41bf4fef : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e217d47431a74b3eb0aace5602413501 : SUCCESS in 11m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9742f0dea5b04c38b87a16e5bdb6b7fa : SUCCESS in 17m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c136e2ffa95842da983427394d7f2fbb : SUCCESS in 8m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8ca2e7df1b67421fba47ae71ef89e82e : SUCCESS in 11m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2d17b2a7640e4b9080c312933a7a0cec : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca54884dd1b04bc7b19ae25351bd8972 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ecb48ab833b478a88bfb7cc99e27129 : SUCCESS in 22m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/aabea6a54241452bb4c7f10b70322c68 : SUCCESS in 6m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9ad50f027f92462cb1e92288fa62213d : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d80dd1a16a4c42159c4a8144841b9b46 : SUCCESS in 6m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/effe2fbfa9c64ee7b85bd3cb292e78b2 : SUCCESS in 13m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2ab2370bf21f4f82940819a47c172b7d : SUCCESS in 4m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b735639d99a24dbaac5d112ee121f768 : SUCCESS in 57m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2a07a32b67c45c1a40bab7aaf7ca7db : SUCCESS in 1h 34m 21s","accounts_in_message":[],"_revision_number":1},{"id":"be0a63da869e7acfe1e43f4280c77cca1be1da78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:12:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46ef93dfc5484ede55c626376ca3511ad888a417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:04:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f363b699619434d954700d5e7ceb77a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e869773db04843efa669d051c850be4d : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6bccfac7e64941817e5759c548ebf9 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa79b45a073d45ec808a8360c1814075 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb4d2d37b0b846639f589840b45d6fc5 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad38ea1171dc4a09a49cd5fe05c32bb1 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bdc4c54f4e64a8ea7c974ff22741de9 : SUCCESS in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5eb7dc45ae4a40d6972ba2ef7f38b5d6 : SUCCESS in 14m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/02cea0c4c9d048859758257cc63c8224 : SUCCESS in 12m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/434590d522b643a8a79a09c36934a0b3 : SUCCESS in 24m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9a3b072957a842d39768fe91f36f2d01 : SUCCESS in 19m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5bed56ac35804e2687daa3b70bdf12b7 : SUCCESS in 11m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1809bb06b8d344fbac6bcbc3e1539c2e : SUCCESS in 4m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2ce8c4178ce44365bdfbad64ed3a2f6f : SUCCESS in 8m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e2866f847e07428290d54f490b5b5346 : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5891b9a473d54998901592be1fe72356 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69268de7b98145e5b4ac1b7b6d8fe4d6 : SUCCESS in 7m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/07204a252c7b4e698456bd9e65699fd4 : SUCCESS in 8m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8410fb3372c44845b9999a6a3ad3f61f : SUCCESS in 14m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/db04dd6d0d5249d2a8179ca65f71b7cd : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ae46be5d886044a685c394e0f1945d4f : SUCCESS in 9m 21s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2b0ab55e15784566a8cc8985d25e81e5 : SUCCESS in 18m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1feeaf9ca0f24ae5a25cacb236910282 : SUCCESS in 9m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/57b1376785ba413aa5c2df24d056fb1f : SUCCESS in 12m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4a8605910f8449f19c2a3af38df2612a : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/271a5db82e264cb8858102e4db052291 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c40841e27354f33a81656673dba9db9 : SUCCESS in 21m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a4f997489b564c5f94350b7aa59f33c8 : SUCCESS in 5m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7d0efc5fc7a845288025d7678af3e525 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc5ab1a2434145e3b8aac9ddf2f6a2f5 : SUCCESS in 6m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fe127ed4c72d419e8fdb01e64b06255f : SUCCESS in 16m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/41b02f70c3c34a5d9610cdbdfdd903b4 : SUCCESS in 5m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5466e10a15464a3484998faa58384d02 : SUCCESS in 58m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eae7c96800d6477d9f08d4a69aeb1b23 : SUCCESS in 1h 50m 36s","accounts_in_message":[],"_revision_number":1},{"id":"805fdb0ee1ba5905008c300e7fcc56dfaa7b18f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:04:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa5ed8adbd86aeb0487663acf33966c74119478b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:05:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1115613b1047bf8bc55fb05bed305a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd517400b0a41f9af40f7a10bb09db3 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62092445df31b29b553ef0499ac19d13cff95817","revisions":{"62092445df31b29b553ef0499ac19d13cff95817":{"kind":"REWORK","_number":1,"created":"2025-05-22 07:32:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/950626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/950626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/950626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/950626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/950626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/950626/1"}}},"commit":{"parents":[{"commit":"0a65df4da38f10e617e584dfcab93403d4587ffa","subject":"add watcher cross job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a65df4da38f10e617e584dfcab93403d4587ffa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-22 07:22:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-22 07:22:39.000000000","tz":0},"subject":"update constraint for python-swiftclient to new release 4.8.0","message":"update constraint for python-swiftclient to new release 4.8.0\n\nmeta: version: 4.8.0\nmeta: team: swift\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Tim Burke \u003ctim.burke@gmail.com\u003e\nmeta: release:Change-Id: Ifa8f3cefdcb392b62d216cb0c69eeda73d3d7a8c\nmeta: release:Code-Review+1: Tim Burke \u003ctburke@nvidia.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I138ee2ddd187be2100579f6a055a0a7d3c7aa973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62092445df31b29b553ef0499ac19d13cff95817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62092445df31b29b553ef0499ac19d13cff95817"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
