)]}'
{"id":"openstack%2Frequirements~950649","triplet_id":"openstack%2Frequirements~master~I727e125e7e232f0570e7b20831b525dc9e5df15d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I727e125e7e232f0570e7b20831b525dc9e5df15d","subject":"update constraint for osc-lib to new release 4.0.2","status":"MERGED","created":"2025-05-22 11:31:33.000000000","updated":"2025-05-22 17:25:47.000000000","submitted":"2025-05-22 17:24:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950649-new-release","meta_rev_id":"ee0c4decc087de92e9a4f625bd272c875f5bb500","_number":950649,"virtual_id_number":950649,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 17:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-05-22 11:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-05-22 15:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-05-22 15:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-22 11:37:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-05-22 13:16:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-22 15:14:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0aa3a76b928cd2ff75ce10704ecbc1043afd0771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-22 11:31:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9e5899133884987bbf6a694f543fb6f03c19980","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-05-22 11:37:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3a602678b9550a59e691bc5f5734ab1fb6c5133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/765ca3461d774bafa656f6e3baddddc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd7952e28e8426a85afb03881d84aef : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abeb2ba7db8c4634b062106c66161b8f : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a05d7ee885c14c5484e22263d1187a59 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c58b972707564787889e90263f70e3af : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b7233b6a9774a23b0c7501e8067c670 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04cb395f357c462293f53a3e2ababfcd : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc6d8957f2874d57b05aae6216b80e72 : SUCCESS in 3m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/22218f6647ac479ebd99f41a0ad2fec0 : SUCCESS in 17m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6b72163b14224118899df256dd1bfd46 : SUCCESS in 12m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8b2fad96b2ee480aa31b326d0d2c970e : SUCCESS in 23m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5e80f1b7f7f448c6a07e709ed67b0149 : SUCCESS in 19m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b9648b94b42a4b52ac4d1025e90544de : SUCCESS in 24m 30s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/efc16023b860463e90460ac820c51521 : SUCCESS in 7m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c4907b8e0e4f468a8c2d6133eaed3d87 : SUCCESS in 6m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/18d1edc2c9b24a4ab4a174e5cb0da50b : SUCCESS in 5m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/aa97d7229f9d4a5d94282909d932c510 : SUCCESS in 12m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c07ea8b41cb04eee957ac300f2e8e9a0 : SUCCESS in 4m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eff6bad95e394c09a6bf9c46844ea48f : SUCCESS in 7m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/899ec17d9c5b4b31a30bb80f7badce38 : SUCCESS in 9m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2ffa3fce3e29494db7d81ee412a09aaf : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3221f6d5723a43e288d430d415ffec76 : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fe4b117b406b405a9a67b68a3cd075a1 : SUCCESS in 9m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2a1b3624eda5470daf5d4a92f446c83b : SUCCESS in 11m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/02ea64a550084c22ae4142151440f608 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/754ac7749dc44261b38528b105f7f78f : SUCCESS in 9m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4d73f882aec24f8fb6d4677e6754adda : SUCCESS in 4m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9a3d69b9ee5c44bf94be1414915de2b2 : SUCCESS in 6m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3f49b097dbc64def98491a09961dc425 : SUCCESS in 10m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e1874dd879a14520a951f2030abc209f : SUCCESS in 16m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8ef67f449e7246a18ca04ec8d27a52df : SUCCESS in 9m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8042f37e3bbf413b9f5de12066c3838f : SUCCESS in 12m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/730773d36bae4cc3a994b523e74ad09e : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1b2875cfe0e45068f0af988b9df013d : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d153b1f90e44fac88537d74ff51c983 : SUCCESS in 26m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8691c9b6d01b400caab43184c5fc699e : SUCCESS in 6m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6e944c2b8f5c4377b6f21fb5ccc029a0 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48178bb5c0cb4a7fa882fd444020e0d7 : SUCCESS in 6m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8f82525f1ecf49afaa6b88fda4ddb787 : SUCCESS in 17m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ed63e144600c404a9136b7b076de2870 : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/755d47c51ad94ee0ba0060982ba7bbb5 : SUCCESS in 57m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3b46c6acb9e44f5a32b3a3f4acad4f7 : SUCCESS in 1h 30m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ac612e91e157a4b99a4f6bad80ec8583e32802a6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-22 15:14:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60d55760095a2b8f74531a6350b225ec48bab686","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 15:15:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af9a8826bcc196fb383928f9b9a7d6afeb8ef532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:24:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476be6ebb48f4810ab6d765eda4c8f75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af37c262545a4ab0a3db621eb50c0052 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3814cad39716436bb40310ed2c3670fc : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad247c4d028442d6b957156292229425 : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69a04676cfad4fa29c8515d74741d788 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0be0e3ea1f44a75bdacb128edd4b31d : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82a8131bf433430db51d101961f68520 : SUCCESS in 4m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bce7520321164d1caf27ecaba882a9f6 : SUCCESS in 16m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c2db0bb9476f44299f3259bee03a4fb5 : SUCCESS in 7m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a5a3fd0d36cb4675bc4db3cf331de68d : SUCCESS in 18m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1301e66bb1824451bc039dad8adddf47 : SUCCESS in 21m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/85fdab96c66b4f8cb9ba7dbd0234aaf3 : SUCCESS in 10m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/50073595a1844d7884c41e57e8b91e5e : SUCCESS in 5m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6ec45244829e4671ad241a83538b7644 : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6f03e4079bb946fe860ac64320066863 : SUCCESS in 7m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2b796b483d504a0c81f8ca48f82d406a : SUCCESS in 13m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c86eb2665fe4c828d0d123669363fc3 : SUCCESS in 8m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/64b0876d028f48a7ae328e08022a4a4f : SUCCESS in 9m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/32ee4dcdf2894dcda57a50a7a49f6425 : SUCCESS in 14m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f78680c046ee4cfe879f17b439b479c4 : SUCCESS in 3m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4178a2332c65442ba7c220235ac42034 : SUCCESS in 11m 51s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/63bcf18c454943678d170260bcf8ef4b : SUCCESS in 17m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2584cbff1dcb4caeb3e7b50b31e1b61c : SUCCESS in 9m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bbf0dffe91784bec96b92627cd3aea47 : SUCCESS in 14m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4289cd3639da4213885091bc2dae9cae : SUCCESS in 6m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa2e2d4ad02c44eba08b16c304a9397f : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41884af92c7142eeb0d64dd52d0d404f : SUCCESS in 25m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/01bfdf1731f740a8aa507be33e6c1572 : SUCCESS in 7m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2fcb91a154f247e0821b1a335dfd8331 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ae0af0ece9c4bd4951bfafb1a950c12 : SUCCESS in 8m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/63b9971d908647c688249bc4937178a5 : SUCCESS in 18m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7d0dc8476dbe4b52a0dbdff831035221 : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b788d58b0b84f57aebe25538f85f4cd : SUCCESS in 52m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b90af82360b428a9cf58a8153c06894 : SUCCESS in 2h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6327020e3dcd9e50a2342db6535a625adc062650","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:24:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee0c4decc087de92e9a4f625bd272c875f5bb500","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4bb5a483414c3a9fd8339b643393b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4611455b276748c786cf2bdc07756d21 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c807c77dfdda351cc8224f17b7b08fbf24624b7","revisions":{"6c807c77dfdda351cc8224f17b7b08fbf24624b7":{"kind":"REWORK","_number":1,"created":"2025-05-22 11:31:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/950649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/950649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/950649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/950649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/950649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/950649/1"}}},"commit":{"parents":[{"commit":"1b4ab48963bec04e6d5c961984b98cf77cf1b637","subject":"Merge \"update constraint for python-swiftclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b4ab48963bec04e6d5c961984b98cf77cf1b637"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-22 11:21:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-22 11:21:32.000000000","tz":0},"subject":"update constraint for osc-lib to new release 4.0.2","message":"update constraint for osc-lib to new release 4.0.2\n\nmeta: version: 4.0.2\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I60a5c383d4671708616bc2bdd156667c3c3b6ea7\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I727e125e7e232f0570e7b20831b525dc9e5df15d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c807c77dfdda351cc8224f17b7b08fbf24624b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c807c77dfdda351cc8224f17b7b08fbf24624b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
