)]}'
{"id":"openstack%2Frequirements~950653","triplet_id":"openstack%2Frequirements~master~I782af5840218d214ba9fdf3ee6a3287c3261793c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I782af5840218d214ba9fdf3ee6a3287c3261793c","subject":"update constraint for oslo.service to new release 4.2.0","status":"MERGED","created":"2025-05-22 12:45:37.000000000","updated":"2025-05-22 16:11:17.000000000","submitted":"2025-05-22 16:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950653-new-release","meta_rev_id":"aa54449cacf54581400ef672f4dda3c8a738f681","_number":950653,"virtual_id_number":950653,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 16:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-05-22 13:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-05-22 13:17:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2025-05-22 14:02:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-05-22 13:25:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-22 13:17:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-05-22 13:25:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-05-22 14:02:48.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-05-22 14:23:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"baa7d0488888f06e438a67d9908ab6204a6d3301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-22 12:45:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d43d780ef9b65941052ac474cbeb23e812298942","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-05-22 13:17:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79f29888fa27ed0f49d60f31cf58649b773fbdb8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-05-22 13:25:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"196e18f6423b4baf1d679a706d65a23ce756b0c5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-05-22 14:02:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a31c0349aacffc1d92248821755bd06cafa263fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:23:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6446598e3064d04abdbeb0ededf89fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf1d3ddae0f4108a6971380e241c96b : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a4d91dfdb58459abac0b121cf947ab5 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6a6301fb58842f2bcf13a0d95ff2646 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16334cf231934be3bd178a79426df5cf : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3acc7fbbe96745fcae0a66669f8e7b47 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/120bb5bf131242aa880291250867abbf : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48c354b94de74e099662d54797193729 : SUCCESS in 2m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8767893485d94ec8ac0f47dccfe5f735 : SUCCESS in 13m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b8676aef867f421089ea6ffe924615d1 : SUCCESS in 8m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f78f2af267164989a813b806b8f93bc3 : SUCCESS in 16m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9753a17c367e430fb2ad0ae549c2bb30 : SUCCESS in 20m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/61652b7691be474787bffef3c8405ca2 : SUCCESS in 24m 30s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4d0f5f623082482091c9fa511560d48f : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d2cf974f7ca24e98bb9bedc5e2a4111b : SUCCESS in 4m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ac32e118aebf4d88a128f0bf8b6c09ea : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e2f94f28f5914c12b5ceea9b4203cdd3 : SUCCESS in 11m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/edab87cc346146de919c9b529bccfabd : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a52a39222da744f0a4d90d2a67e835a8 : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8b3c1a823afe4341b185910422291a17 : SUCCESS in 7m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/460927dcef5a4de4947f7ddd09d13753 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36b8a1beed89446abe21909ff7a114a7 : SUCCESS in 5m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7588719a4e5a4f8580ce940225e8ad13 : SUCCESS in 7m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7edf00624e4f4fbe957089c89cdb5755 : SUCCESS in 7m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0b29d887d7dd4023b64592f30d4a63c4 : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ca34f69d695b49ebbd700ea1fa59a5cd : SUCCESS in 12m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/31b0d4531d6f4060bd8c9b9b29ab1a46 : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4d51fa4fad43416886288f62ca5d2c73 : SUCCESS in 4m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6c96a48f4d2e4962a469ecd7edcc7d07 : SUCCESS in 11m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2b0001d728e54b56bbde2b8d73b90d80 : SUCCESS in 18m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1eaa30a0ee434fc3a3aa99d1a56977ab : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cd4fa03a03a643b6873e2ae9e6cbb0c0 : SUCCESS in 11m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/96e00f7741df47b1bdfbc3f7b1d018ce : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/76b9740367bc4660b759bb1ec242c94b : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96354a31c4284607af2e7f740ee27ad3 : SUCCESS in 23m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f5b93ab00332430a82eddf3b668bc4ce : SUCCESS in 5m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a3840ebebbb04bb0a7735c8f371c5134 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d552f9a7c554f0d8b50329d956f12e8 : SUCCESS in 6m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1e2264dc6a5d406e9de5715eb6d75661 : SUCCESS in 17m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/153f3685964c4704bd8fdc0fadceb9d4 : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aad9e622bb6f4b628b967a95d24970d6 : SUCCESS in 56m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed9b7c091d894b319dde87a2e43a077c : SUCCESS in 1h 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fcafcfb37daf2bf76241de28c2b95fd46d61cb41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:24:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18d9a09a726a0e01df72494fc1e61e1f0bc2c6be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 16:10:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa3731588094d779d6645f406d6d9c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8512d61ea4c64c3ea121bc26c4977c33 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528c5973e38d4ca9a47787a85cd8ad32 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca13227df0a14cadab89b792c5abc3f3 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc97781a8fb6484ea3f5fc610743e7fb : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92e967d39b9a41b3a6da92f69d768a07 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f72fd241149e4b2e97597eb0252cc10b : SUCCESS in 3m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69b313a4727d4e9eb167c01f2fb5282f : SUCCESS in 18m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/368bb23aab9244e090f1a9cf69fab49e : SUCCESS in 9m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/87cc515788664dbba58590b31c6cc7e5 : SUCCESS in 17m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3455001dbc45470eb51baae1c1286bc7 : SUCCESS in 19m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1083a46580ff49d4b4a83a6acc9b4d4c : SUCCESS in 14m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/99c4596a0c414106a4f63e5b3c31583d : SUCCESS in 4m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fa198a4ab9a64f34af88ca0b4910485a : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e5f0a9970d59448cb2d7ff7b1877540a : SUCCESS in 10m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a596245b1fcc49bc848f019e2ee03d8c : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5421fbe92a0b48cda2e5f9ac59422715 : SUCCESS in 8m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/af7e989fd72143889cac9745b5a6822b : SUCCESS in 8m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/09b09a56238841b8a8dc60e74659e038 : SUCCESS in 13m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/58ca9ff0e15b4475b2564b0f6b63cca5 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b66d81afb71b4dddb8fbdd00d7f50ab1 : SUCCESS in 10m 40s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e0d8d61f263e4625aadc980fe35a55cc : SUCCESS in 20m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/639ddc2466814b94b49a1ea18a0f2532 : SUCCESS in 9m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/53584ee4630b46dca05fb8099a57fff3 : SUCCESS in 13m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a57de8c472e24267954b7651ec0aeeb0 : SUCCESS in 6m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1714674bdea145dcb79bb55391eb3476 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b00558d1215d4fbf8481698855436965 : SUCCESS in 25m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e2bcab1f35954e1f911b81e75bcb2987 : SUCCESS in 7m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c8688f841dc5444a99397138cd6af90b : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e45c7a897f2443d9b3bad8cf30e37b55 : SUCCESS in 3m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/75ba97ca0a314d1d96ef88d112df8603 : SUCCESS in 21m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6dd334205cb14e838ce5653fb4fbb3d3 : SUCCESS in 4m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2409d1b006c04c9da4d6fa2c910e0a7e : SUCCESS in 1h 04m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/076711a8e7f64397825c5e2abde7f217 : SUCCESS in 1h 40m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8f12213303644f952a74f1e3e23396f8519be23b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 16:10:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa54449cacf54581400ef672f4dda3c8a738f681","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 16:11:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d2eb823e93841c3af61f80ed3c37074\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905f9e30d5ef4285a22216d81fc8bb5c : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28c6e98e689cbae9faf0fe5bbb900e074f12c4b9","revisions":{"28c6e98e689cbae9faf0fe5bbb900e074f12c4b9":{"kind":"REWORK","_number":1,"created":"2025-05-22 12:45:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/950653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/950653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/950653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/950653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/950653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/950653/1"}}},"commit":{"parents":[{"commit":"1b4ab48963bec04e6d5c961984b98cf77cf1b637","subject":"Merge \"update constraint for python-swiftclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b4ab48963bec04e6d5c961984b98cf77cf1b637"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-22 12:35:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-22 12:35:35.000000000","tz":0},"subject":"update constraint for oslo.service to new release 4.2.0","message":"update constraint for oslo.service to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ie4603517edd610f7fd65d8629313e7a1c8644a30\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I782af5840218d214ba9fdf3ee6a3287c3261793c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28c6e98e689cbae9faf0fe5bbb900e074f12c4b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28c6e98e689cbae9faf0fe5bbb900e074f12c4b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
