)]}'
{"id":"openstack%2Frequirements~950734","triplet_id":"openstack%2Frequirements~master~Ib7dc3de96598aa21cc59fa2daf4ace95afaa8b9f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ib7dc3de96598aa21cc59fa2daf4ace95afaa8b9f","subject":"DNM: split u-c bump part 1","status":"ABANDONED","created":"2025-05-23 08:07:41.000000000","updated":"2025-05-27 09:56:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70d59ce6cea9cc091add3d905e7558d03f115c73","_number":950734,"virtual_id_number":950734,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-23 11:07:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 11:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d22ef3aa111738be0cffa560df9d4d067c8476c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 08:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aac38ff9fffabb1c7f1b31e0b326b3e0d5e167f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 09:15:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e4c494e341a9e15ed8d74b72f7061f1d35718523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 11:07:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29fdbbcce4844418bc158fbd565956ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50931a0a57174c7c96b56f1d21d0964f : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f8383d176a4e07955d22b6fb1e8ac2 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9030750e30754701ab3a3b5bddf7e3bb : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/688e852c6c7b4767823fa5c3b49b79ca : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da391fd630eb419e9672016f22808372 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4fd56dcad214359b6d91c16a2f43eaf : SUCCESS in 1m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50ae149bc169483e999bef0412a8f9ff : SUCCESS in 2m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/842efb9ce8464cdb99e8cededc5d7443 : SUCCESS in 15m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/544270cbc3ac4b73a4ad4ad5bc7d648e : SUCCESS in 8m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f2f7da27c32c43f2ba82b47c62202a33 : SUCCESS in 18m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/394f185e070d4fc08cd55c312fac9bc0 : SUCCESS in 19m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/717210cf81ad40cdbbe1d619b496148e : SUCCESS in 25m 47s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5a4836dd85d3455594da8f3b593ba968 : SUCCESS in 4m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a2fc70b21aa043628d07714e510326d3 : SUCCESS in 2m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d954f6460244417b9e42a1c4c7f1a1aa : SUCCESS in 2m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/447c5e0a43854c95ab9b5b38013f6632 : SUCCESS in 12m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fe13586c22a24942938bcfee5d55d2b4 : SUCCESS in 4m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/84f9e8a026934364ae2d1ba4023c5e70 : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/899ee03e3c7347b2bc38c1b7745a6af8 : SUCCESS in 6m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0003ddc95cb442a49154fe0cf7eb043f : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c92e173386bc4957af9bcae33144e127 : SUCCESS in 6m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c3ebe0c406d247eca785e43a4dd40519 : SUCCESS in 5m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/14985525b28c40e18dc190cbcb88f6a5 : SUCCESS in 12m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/27a556e855aa4c55833309fb6bdb3b4a : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7897eeeee3fe4e00afcdc73aeb6e52f6 : SUCCESS in 11m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7b75c4207dbe4aef908a96a2120eb300 : SUCCESS in 6m 32s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9dc053cac8ce4666928ca34add9b6a09 : SUCCESS in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3245c793a45a4fc0960d7320ce2f73f1 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/74796d90bea44ecf85496b85ef60ecc9 : SUCCESS in 17m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1aefe1b8943f46bb89630465b3d2078d : SUCCESS in 8m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4a060360a9284632ab72ba29c9fe39cf : SUCCESS in 13m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ef85075303754c9d93032958911d3c6d : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3083a12ffd514adf9ae825cd78fc0554 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d69c6983d7a41fdb20f4945d2e820be : SUCCESS in 25m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a6d091a6d7e34926a283d8e3b638fd4c : SUCCESS in 5m 59s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/84c8996b3ca345c9a535d63f5de37d7e : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cab3c25de5048998011e28f033cc166 : SUCCESS in 5m 26s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/71bfb0e960de4b1bb5a5404f993bdeb1 : SUCCESS in 16m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4af51f1c1b454fe9a1c542ffb4877306 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1472572bb4e4dae8592397f1a164743 : SUCCESS in 57m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8abb11a82b8646af8cdd4a2c565b4809 : SUCCESS in 1h 38m 51s","accounts_in_message":[],"_revision_number":2},{"id":"70d59ce6cea9cc091add3d905e7558d03f115c73","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 09:56:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5373420496fa283313bc6b5ff5c3bf1405e6b556","revisions":{"9e0fdcf1ac2fd86fbc516baf5a021d12e67c766e":{"kind":"REWORK","_number":1,"created":"2025-05-23 08:07:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/34/950734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/950734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/950734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/950734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/950734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/950734/1"}}},"commit":{"parents":[{"commit":"d43f508ec1b3383a35b9d2299d36add1d2521a62","subject":"Merge \"update constraint for osc-lib to new release 4.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d43f508ec1b3383a35b9d2299d36add1d2521a62"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:04:52.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:04:52.000000000","tz":120},"subject":"DNM: split u-c bump part 1","message":"DNM: split u-c bump part 1\n\nChange-Id: Ib7dc3de96598aa21cc59fa2daf4ace95afaa8b9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e0fdcf1ac2fd86fbc516baf5a021d12e67c766e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e0fdcf1ac2fd86fbc516baf5a021d12e67c766e"}]},"branch":"refs/heads/master"},"5373420496fa283313bc6b5ff5c3bf1405e6b556":{"kind":"REWORK","_number":2,"created":"2025-05-23 09:15:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/34/950734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/950734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/950734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/950734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/950734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/950734/2"}}},"commit":{"parents":[{"commit":"d43f508ec1b3383a35b9d2299d36add1d2521a62","subject":"Merge \"update constraint for osc-lib to new release 4.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d43f508ec1b3383a35b9d2299d36add1d2521a62"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:04:52.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 09:12:34.000000000","tz":120},"subject":"DNM: split u-c bump part 1","message":"DNM: split u-c bump part 1\n\nChange-Id: Ib7dc3de96598aa21cc59fa2daf4ace95afaa8b9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5373420496fa283313bc6b5ff5c3bf1405e6b556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5373420496fa283313bc6b5ff5c3bf1405e6b556"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
