)]}'
{"id":"openstack%2Frequirements~950735","triplet_id":"openstack%2Frequirements~master~I12c7a1d3519c2cc554bf2eaee8272783ee222641","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I12c7a1d3519c2cc554bf2eaee8272783ee222641","subject":"DNM: split u-c bump part 2","status":"ABANDONED","created":"2025-05-23 08:07:41.000000000","updated":"2025-05-27 09:56:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce84a6e496919067afcc17e83824065e73f8180e","_number":950735,"virtual_id_number":950735,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-23 11:08:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 11:08:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74b50f26dcb36a08c3c6334a6edc2a72857c460a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 08:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93e501658da7eec4e76bce73eb0409a6adc92d4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 09:15:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f2559cd672870f694e7f0988ff7d8cfbeaf09188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 11:08:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be78ad3f75f04740805f8aaf146096f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495ae77106264184ab3f207224f15279 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd9d96121c140a28d2a0955fa2540e1 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11cf5099ae064ef29abdb58efa4eb364 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19e3c33c44dd472285b1a1065a0d80e1 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0e1c055dfe547c8ae7cd2e83e7bef6c : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e97fdb90399b4d8c98a712f26607b630 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d6144f248a34a4099be6820d4e0f431 : SUCCESS in 3m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8d6388648515468ba943cc02a915b3b4 : SUCCESS in 14m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f132b957432241c0b64ce662696df734 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d93d182ef75e43c2b696bcee025c23a2 : SUCCESS in 18m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/14f97e66dc04450fa6a4e030bba32e42 : SUCCESS in 24m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/67267682cdde48c4947032ce83fa2c92 : SUCCESS in 19m 19s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c73f9744cb244560bec2ec89112d7e2c : SUCCESS in 7m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/38730fe468e34d56820f0d9b876889fb : SUCCESS in 5m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6999023bee864b4dae0bf6f98b4c2ce2 : SUCCESS in 3m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e07b48e7c7794e498389c2c346cce236 : SUCCESS in 12m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e980043a30724a5586afdb8d074645b5 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3cae2c5e51504bcab67287cfc4c18176 : SUCCESS in 6m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/139e4d9f9dd74f9b82c05607f6aeebcc : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5de2f8c85d2343b29b6c0e96f9a032db : SUCCESS in 14m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cdfba75267f5450a929904826b0d1527 : SUCCESS in 5m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b26b01e0519d4bda9e1b4d494288fde9 : SUCCESS in 10m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/91806c8a691941ebaab1a4425389bed4 : SUCCESS in 11m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d8e3c23695ee4d71b38fb05283a04991 : SUCCESS in 4m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/45980da0c1584a24b4b10a423cf78c3f : SUCCESS in 10m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/cf5f732f859145419251127210afd7c0 : SUCCESS in 6m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6b70dea6f00f4ea7816c8e7b30bea09d : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0b0fb1f4f868453d8c2d50e2e2531168 : SUCCESS in 10m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a42715d694bd4bf788e18371c9dca21d : SUCCESS in 21m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e4eba76310d0456983f5766433217ed7 : SUCCESS in 9m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/48fa6b705a1e499f8c4388e235b51575 : SUCCESS in 11m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/28bec89060df42bcaafc11e5480d274b : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d54c61b1ec7c41c88a40b644d7c6b377 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01730590a54142b9ad3797e008490d2e : SUCCESS in 22m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8b6bf2491ec04cceac7c1834d5e14b1c : SUCCESS in 7m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5828a3ed7b35430b96c4f4396b862c1f : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f09c00e6445b4b2da116970d91d9cdf4 : SUCCESS in 6m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5c3f2788546e4869826e6640aa590ff2 : SUCCESS in 13m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f599348c3d334400b74ef126d1417b58 : SUCCESS in 6m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dd9f7b4d66c49ffba1b4fca53f3f6c3 : SUCCESS in 52m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3a048abad6442f5a7b2561de09a8588 : SUCCESS in 1h 48m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ce84a6e496919067afcc17e83824065e73f8180e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 09:56:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c2e800c1ea2fe413dfcea51a739dc94b8e752543","revisions":{"bc24e23c260aa2209403594d376aadfadf484e48":{"kind":"REWORK","_number":1,"created":"2025-05-23 08:07:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/950735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/950735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/950735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/950735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/950735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/950735/1"}}},"commit":{"parents":[{"commit":"9e0fdcf1ac2fd86fbc516baf5a021d12e67c766e","subject":"DNM: split u-c bump part 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e0fdcf1ac2fd86fbc516baf5a021d12e67c766e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:05:23.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:05:23.000000000","tz":120},"subject":"DNM: split u-c bump part 2","message":"DNM: split u-c bump part 2\n\nChange-Id: I12c7a1d3519c2cc554bf2eaee8272783ee222641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc24e23c260aa2209403594d376aadfadf484e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc24e23c260aa2209403594d376aadfadf484e48"}]},"branch":"refs/heads/master"},"c2e800c1ea2fe413dfcea51a739dc94b8e752543":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-23 09:15:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/950735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/950735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/950735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/950735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/950735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/950735/2"}}},"commit":{"parents":[{"commit":"5373420496fa283313bc6b5ff5c3bf1405e6b556","subject":"DNM: split u-c bump part 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5373420496fa283313bc6b5ff5c3bf1405e6b556"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:05:23.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 09:13:14.000000000","tz":120},"subject":"DNM: split u-c bump part 2","message":"DNM: split u-c bump part 2\n\nChange-Id: I12c7a1d3519c2cc554bf2eaee8272783ee222641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2e800c1ea2fe413dfcea51a739dc94b8e752543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2e800c1ea2fe413dfcea51a739dc94b8e752543"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
