)]}'
{"id":"openstack%2Frequirements~950736","triplet_id":"openstack%2Frequirements~master~I0d504516e40a7c4ece2e654f9bc40e7b67e7286a","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-24 08:33:48.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I0d504516e40a7c4ece2e654f9bc40e7b67e7286a","subject":"DNM: split u-c bump part 3","status":"ABANDONED","created":"2025-05-23 08:07:41.000000000","updated":"2025-05-24 13:35:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5f2fc57bef0ccb6a1bbd68f303a644e86d99e09","_number":950736,"virtual_id_number":950736,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-23 09:51:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 08:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d781523be3e8b802fd52d11c1b8fd9d1af62deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 08:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cabcf6694aac5415ec618ded08f858a7f00f98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 08:59:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8a25a9e909944269f9417ce7c031f50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858af35244d04ce29e80a45b3506a679 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086b0ea1c6a54a73b2f6f5cd1af983cf : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e9d4547f8dc4e64aefe0c4592f3fe53 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/118e17bcb0154c5792f45c3ebec49657 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c32629ddfa8045648c4b4ac800d2da94 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/932f87a3fe2146a181bef456296ead5c : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26535fff831a4d5f961afd69130e4d73 : SUCCESS in 2m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9018e364cbcc4f9fab387ac4bfa6a095 : FAILURE in 5m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/767f1d8d5af845acb93e816e2785d7be : FAILURE in 3m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/42be1a6173e1455aafc455f44e4a12db : FAILURE in 7m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9ac1e251455445dcac9f706931e34d09 : FAILURE in 13m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dafb261ed8c34c4196857a72e8d813ce : FAILURE in 16m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e5691968ff934db1aa7708f3231a48f8 : FAILURE in 4m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3eeb44f7c9b5435e8eb49b9262243e98 : SUCCESS in 6m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/05833605e26444aa996c265a707cb542 : SUCCESS in 6m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/89c8ec698ea24e8aaa3f72c54d8fdeac : FAILURE in 4m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d27c8b95534242f488efd2d1d4d57a72 : SUCCESS in 4m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/462ade09dc314abab6d53ff6baabfeee : FAILURE in 3m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a65fc823b257430cb77316f472790d21 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3beb04b5eb8141eab1131110c3c56580 : SUCCESS in 13m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ec77e4996b94bf8ae708edd85265acb : SUCCESS in 8m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9c3a4a606e8a4eee9e18745ac88c5b3d : SUCCESS in 7m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/adc4299291e144f79152b0f43576fb62 : SUCCESS in 12m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/698d80ad55a842f6a99816888aac464b : SUCCESS in 2m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9c5d30efa1574a46a2eaf071feb3dc18 : SUCCESS in 9m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/81557542ad5e410fb0db901b21f7a3b0 : SUCCESS in 3m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ad341986320243579e70b93afc80ec20 : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9472bf46e2ec4b6aade8ef1316ff2a44 : SUCCESS in 11m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6f1234238ac54cf8be09b95be35203ab : SUCCESS in 19m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a5825817a23648fc91023a3c4d59638f : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/809535ea2b6a47cab9b118562da86fb3 : SUCCESS in 12m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6ce59f5bc4314eb9bc043148d429874c : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1716b182905e422884db2c8de4a2d14e : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33591e1ff7d44802bdc138ca8b0d2433 : SUCCESS in 26m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/64838eba754a4fcaa0bbb62eef868fa4 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bfda8e67154e40e1ac2aa145f114bf59 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/089b332e351a4f3ebf8a8f604f23cec0 : SUCCESS in 6m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ad3729075f1246539e37f82899afa8af : FAILURE in 3m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7a9df4ad669246d0a6bec6a11d6d3140 : SUCCESS in 5m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97b07cbbb36a4808a64b7f92a3810bbd : POST_FAILURE in 10m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e78d8f9885d43b4883a774ecbcd0094 : FAILURE in 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"97420af2546b45a9a11712fcbb310bb153add604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 09:15:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"16c21be1058bd757e262198420ffc1ac482d5fc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 09:51:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/608513d2459a45a485ddaeabd59ef33d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b352aa7e2f0e4ad79629c4a4f214e9c6 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55801c883c2b418baf018bf6af687854 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87a24533719e4d2facb43d9de9f06a0b : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee349be5540d44948f37da75e666135d : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b5e79230a514714b0a497666e677907 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa7a94b7a2b6487593b7d5e8507add0f : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ed1f1be81644d4da4babff4ad135aac : SUCCESS in 4m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9d57fd1442ad40a6a6d79835cc7a24f4 : FAILURE in 6m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b73f194346b74a279e45f6f0389a9cd0 : FAILURE in 4m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/89ea0fc143874fa9aadaab2f942f348a : FAILURE in 13m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a73ffa1c533e4c51917d191d9d8ea042 : FAILURE in 7m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e6898fbc0ebb4c5d8f9de7abef8e1d55 : FAILURE in 11m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1cc1728c0fd64f5f85f4872abe2d1bdb : SUCCESS in 7m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7a3ad5e42e1b47ed86882fed0f19c495 : SUCCESS in 5m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1e9c137142dd4cfd971a51b11276e861 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cccc280ba6ce4941a2ca608aa6e3baac : FAILURE in 5m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bd221fc95ab74b4f9eea660c61c1f9bc : SUCCESS in 6m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/69162dbed6734660ad9e981845714c60 : FAILURE in 4m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9d7311ea6ef94d3eb96fa57f4b0ec90d : SUCCESS in 7m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/30bd993e58374af4b83f38e14723cb81 : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34dfbe65684240bc887553cbf9c539e3 : SUCCESS in 5m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/061affae73324efc96b253494291faf2 : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8a3a7646a82d4a838e2116b5001658d8 : SUCCESS in 10m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6fbf2a3d1be642139211027d488cd5d4 : SUCCESS in 3m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b3b7ccc9108b4464baf868dcc557ad17 : SUCCESS in 10m 41s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/aff25cbca0e84994b3ca3e1295b75a7a : SUCCESS in 5m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/66310827a5de4c0e9811dfa30d5321bf : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0f33e605c17f4cd5aaa99ac4eeeec17b : SUCCESS in 12m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/af901946355046578a2c6714b8febcc2 : SUCCESS in 20m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8a72d0e452447889f48d3a18b6c2e2f : SUCCESS in 9m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3a42a2f8918e4b5bbe8faef8dc00211b : SUCCESS in 12m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3c00b78431cb46eb9181c78fca11db78 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f1cb853ee1504495b42c15539a0848db : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd62c5eef6fa46e288b25a66c3e97ff7 : SUCCESS in 25m 31s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/253fb5ffcbf74b4788966b1063abbf12 : SUCCESS in 5m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0fea507cb1bd4fb5b72a2dd3677e5d11 : SUCCESS in 3m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73d1b18dabc44ba6a6e7ec5599b744d2 : SUCCESS in 6m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b88e20b938404dc5b84de665c0718db1 : FAILURE in 2m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d709347d00a145a1a5c81a0442667a36 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d1b1079e6624e00a95446d819b7c524 : POST_FAILURE in 13m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2296e1e0ad9d44d6bcd6a396c9b68261 : FAILURE in 11m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d5f2fc57bef0ccb6a1bbd68f303a644e86d99e09","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea6833049cd9b127840d25982e90eb807435f892","revisions":{"bc8335e948c701ae617d9b1f47dfd9f8bd1dc8dd":{"kind":"REWORK","_number":1,"created":"2025-05-23 08:07:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/36/950736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/950736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/950736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/950736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/950736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/950736/1"}}},"commit":{"parents":[{"commit":"bc24e23c260aa2209403594d376aadfadf484e48","subject":"DNM: split u-c bump part 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc24e23c260aa2209403594d376aadfadf484e48"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:06:01.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:06:01.000000000","tz":120},"subject":"DNM: split u-c bump part 3","message":"DNM: split u-c bump part 3\n\nChange-Id: I0d504516e40a7c4ece2e654f9bc40e7b67e7286a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc8335e948c701ae617d9b1f47dfd9f8bd1dc8dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc8335e948c701ae617d9b1f47dfd9f8bd1dc8dd"}]},"branch":"refs/heads/master"},"ea6833049cd9b127840d25982e90eb807435f892":{"kind":"REWORK","_number":2,"created":"2025-05-23 09:15:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/36/950736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/950736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/950736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/950736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/950736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/950736/2"}}},"commit":{"parents":[{"commit":"c2e800c1ea2fe413dfcea51a739dc94b8e752543","subject":"DNM: split u-c bump part 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2e800c1ea2fe413dfcea51a739dc94b8e752543"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:06:01.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 09:13:46.000000000","tz":120},"subject":"DNM: split u-c bump part 3","message":"DNM: split u-c bump part 3\n\nChange-Id: I0d504516e40a7c4ece2e654f9bc40e7b67e7286a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea6833049cd9b127840d25982e90eb807435f892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea6833049cd9b127840d25982e90eb807435f892"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
