)]}'
{"id":"openstack%2Frequirements~950738","triplet_id":"openstack%2Frequirements~master~I4da15673d1b01c2e44b918d2b53de9edd14b2657","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-24 08:33:51.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I4da15673d1b01c2e44b918d2b53de9edd14b2657","subject":"DNM: split u-c bump part 5","status":"ABANDONED","created":"2025-05-23 08:07:41.000000000","updated":"2025-05-24 13:35:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa7519932dbea4cf1ba967318f8fbc9a6719ae18","_number":950738,"virtual_id_number":950738,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-23 10:55:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 10:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8ef364d2971fea4898b50813a8df5eaae11839e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 08:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ba4814ede18216cb40b404c258c8acf8e5b6519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 09:15:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2c8ac85f2340c662be2e1e920dcf4c24bee84760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 10:55:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45be16c826934860a14db29614cddd9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9147014d7e491c9bed131c79530f44 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a7aee5dc814674a147c13cceab0f35 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cd44126836c4014a4189bdcc1b9f84a : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9427891ba2e42d1a484674d38c8a0f4 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/083c943eefe847078d7505fc5c4d1e6b : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a8ee8deee0a4a709439d15373a485f3 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7509f6a029ac41d59b50e0d179001491 : SUCCESS in 4m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6c1746bb023d49388b27f2e02f2d285d : FAILURE in 17m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/38e014c379a14d4cbe20b9142289b191 : FAILURE in 9m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/38b3bc0f93aa4fcd99c39d58f15ab5a0 : FAILURE in 19m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/acc58c98616349ba88f30d74ee97cb2b : FAILURE in 18m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9269e4d698d14f5e96be46235ee1afdc : FAILURE in 27m 30s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f89a731e0ef24ee2b10ac1d364edb227 : SUCCESS in 4m 50s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5a956691b2734e42bed49bf0f9275960 : SUCCESS in 5m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b911591d441a4bc49854b054bf5e2f39 : SUCCESS in 4m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fe09ea4620134a159bbc6a532d8dfa14 : SUCCESS in 13m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f7ae9eab69bf44059b0678f584da616b : SUCCESS in 4m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/db093daf27f24e3892df89b4ed2ee51b : SUCCESS in 7m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/05f65dca318b495f8c6becd707b46025 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b73da37a52d948778d4d40dc31d9495c : SUCCESS in 14m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86396fc2976e4703a586fc89f4d3d68c : SUCCESS in 6m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2c42f9e5f72643ec8c9f38b1b6c043d7 : SUCCESS in 7m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/38fc079751064506906681d4e0352c82 : SUCCESS in 13m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0f26369671474903a1ca575c1dc17087 : SUCCESS in 3m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5759e1b167b84c73a4ee576d7423be55 : SUCCESS in 11m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/342ca1f939df44a589e4bc69fa02a712 : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3e4e61dbb7e04aeb8f1fa37cf330d1dc : SUCCESS in 5m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7ecc00e2b9e94d5fa928939a3100a4a0 : SUCCESS in 11m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e9f06e2c3af54b589ca3cc59bdafeaef : SUCCESS in 15m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/67091dc3e36e4801b83e69dc2eb04091 : SUCCESS in 7m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a8d084982a524a0aba7ce09b7bebdbfd : SUCCESS in 7m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/436776334d3a4f57b6559a38083c80fb : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b63257b01442444da1ed49af24e05f18 : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8eb2e64f9e247039551ab11ed9fde50 : SUCCESS in 27m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/21f7ec9bfc2f4e1f9e244d8553a75f79 : SUCCESS in 7m 59s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a392af37d8794df996525ece752c4da5 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bba9cc0411df4d50844ee3f33df8f0ae : SUCCESS in 7m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4db720420dcc4f78a4bb67d2d48e22c4 : SUCCESS in 11m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0cb25dfbbe154862ac6f1113f081ba41 : SUCCESS in 7m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d90782bc6fb4cb1b28032719363caa6 : SUCCESS in 51m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5565c5396ead4d2f983f57d59dd80673 : SUCCESS in 1h 27m 29s","accounts_in_message":[],"_revision_number":2},{"id":"aa7519932dbea4cf1ba967318f8fbc9a6719ae18","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c445c24ca3271f4b263914310e62e39304e91fa","revisions":{"fe2d547198833ba8c2801d86a0244ababd5f55e4":{"kind":"REWORK","_number":1,"created":"2025-05-23 08:07:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/38/950738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/950738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/950738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/950738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/950738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/950738/1"}}},"commit":{"parents":[{"commit":"3db1ec7b420cfc5611c6311a6fc1f3029edc80a6","subject":"DNM: split u-c bump part 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3db1ec7b420cfc5611c6311a6fc1f3029edc80a6"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:07:19.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:07:19.000000000","tz":120},"subject":"DNM: split u-c bump part 5","message":"DNM: split u-c bump part 5\n\nChange-Id: I4da15673d1b01c2e44b918d2b53de9edd14b2657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe2d547198833ba8c2801d86a0244ababd5f55e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe2d547198833ba8c2801d86a0244ababd5f55e4"}]},"branch":"refs/heads/master"},"9c445c24ca3271f4b263914310e62e39304e91fa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-23 09:15:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/38/950738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/950738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/950738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/950738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/950738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/950738/2"}}},"commit":{"parents":[{"commit":"41e62de91667b1c1b7cf4a98d79af7df4b2abea9","subject":"DNM: split u-c bump part 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41e62de91667b1c1b7cf4a98d79af7df4b2abea9"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 08:07:19.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 09:14:12.000000000","tz":120},"subject":"DNM: split u-c bump part 5","message":"DNM: split u-c bump part 5\n\nChange-Id: I4da15673d1b01c2e44b918d2b53de9edd14b2657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c445c24ca3271f4b263914310e62e39304e91fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c445c24ca3271f4b263914310e62e39304e91fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
