)]}'
{"id":"openstack%2Frequirements~950744","triplet_id":"openstack%2Frequirements~master~I8af7da004b54f5d95d7197a1db1484302a0cda02","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I8af7da004b54f5d95d7197a1db1484302a0cda02","subject":"DNM: split u-c bump part 3a","status":"ABANDONED","created":"2025-05-23 11:06:58.000000000","updated":"2025-05-27 09:56:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"322a2f88db2ce99ea10f54f474fb8a269a856a2c","_number":950744,"virtual_id_number":950744,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-23 12:45:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 12:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41ac2f1a754b2555f091f0624a16b8d5e92740a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 11:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4350a0f2040246592bee83e5d80dd543c321ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 12:45:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecbeda1963264b1e9f1496606c2d8ece\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30e322540a8493abe40ce2faabf473e : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c3146d259a489db3723b9fbe6b86f9 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/857313a083bb4566bbcc0585dc6a7422 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e1d5669e48f41dfa6911ef62485601e : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9804677924774be8a9a52ec14930e2ed : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e93fbf112aa444c85b6c677e3a452f1 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be5c84ffaa1a4cd8b39eb2ed97abb6fd : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/61f1334f81524714a7e409e920a5a068 : SUCCESS in 17m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/35dff4c8abd643fc83902bd7490911fe : SUCCESS in 9m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ddfb7d98a4e34498bb22b92cb8d0f630 : SUCCESS in 18m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/331b6542024b43db908e2b06e8b2b55a : SUCCESS in 19m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e2dc4f9646e342b48286b5ad41d72cca : SUCCESS in 25m 05s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/130db6dd4edb49afac046474b07c2295 : SUCCESS in 4m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1179aa1b7bf34c9dbb6c5d89325e4187 : SUCCESS in 5m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4d58f21b25b84415af8164d50cac7bcc : SUCCESS in 3m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/45e37823340e43b497829531730365bb : SUCCESS in 12m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/593911639f514114af08635f383d99ab : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0b1b1005e6374933b9a412bc8960d939 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2614cb87815c4361a20f72ee923ca369 : SUCCESS in 7m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/554692c2f00342a486936da3ff5ba6da : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1719a06b7bd4c3db00082eeee81aa0c : SUCCESS in 7m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/971362fdff1a4f3493e857b7131fbcc7 : SUCCESS in 7m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/abcc01c3c93248bdaa99848118928eff : SUCCESS in 12m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3e6b6ca0202b4bc38b884e8459d37a7e : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1a6adc65d76440f9b6c4379c6c4d718a : SUCCESS in 12m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7119fb1e26d546cabcee9473b24edb7a : SUCCESS in 4m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/53d2761fde0c4fb99d8108cc1a07e080 : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fe42442f573641b488b82306c6796fa1 : SUCCESS in 11m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2ae93d49671a4c34b307aeabd778e3c5 : SUCCESS in 17m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/47d310708c6546df8039ea07f545a2f7 : SUCCESS in 8m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/809382e30bc0457ba21bd3995b007737 : SUCCESS in 13m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/85a606f250064f78af1c37c4367b011a : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f743d328c694b71aa19df6be20ae7d4 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f2a62aac92934866bd9f664fc5474deb : SUCCESS in 21m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bbdec86160704ce98d648566dbbf7161 : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b6514457ae5c492396415b1bf076700f : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/954a135ab0d146b6b75d43bf4e45743c : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3c9908b9cc0b40f08a909dd36fdf6c95 : SUCCESS in 16m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5acd4dedc0e245a8878dac28d0a2d5e0 : SUCCESS in 3m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8b8ca31fe92444ab65f95f595b7e3c8 : SUCCESS in 56m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80d307b0adc44c0b8f39c3537d55f490 : SUCCESS in 1h 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"322a2f88db2ce99ea10f54f474fb8a269a856a2c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 09:56:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a11be4ff80ae6666274712107b629a29af77e2c","revisions":{"4a11be4ff80ae6666274712107b629a29af77e2c":{"kind":"REWORK","_number":1,"created":"2025-05-23 11:06:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/44/950744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/950744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/950744/1"}}},"commit":{"parents":[{"commit":"9df4c3b73418755140fe442a389552bccaa6ea75","subject":"Merge \"Deprecate openstack/openstack-ansible-tests repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df4c3b73418755140fe442a389552bccaa6ea75"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:30.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:30.000000000","tz":120},"subject":"DNM: split u-c bump part 3a","message":"DNM: split u-c bump part 3a\n\nChange-Id: I8af7da004b54f5d95d7197a1db1484302a0cda02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a11be4ff80ae6666274712107b629a29af77e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a11be4ff80ae6666274712107b629a29af77e2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
