)]}'
{"id":"openstack%2Frequirements~950745","triplet_id":"openstack%2Frequirements~master~I0f311f73a35ba6d7211d92997b4047aef9524942","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I0f311f73a35ba6d7211d92997b4047aef9524942","subject":"DNM: split u-c bump part 3b","status":"ABANDONED","created":"2025-05-23 11:06:58.000000000","updated":"2025-05-27 09:56:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9967d4bc0a9e5fc3f57e9611936a06c7ce2bf1b3","_number":950745,"virtual_id_number":950745,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-23 13:03:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 13:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13a9b148a1ae1b8eb3956121d61be4980999a964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 11:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f7a0a8af71095f01b10a7c6cd99d2cac3717f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 13:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e9bbdd14b04740ac50bae9634972ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/539af96e26f14982bd120c4df7fb1c41 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1083209c1940ad983d366f8ffdc131 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb10cffcccbc4bf3a593e5a671542f0e : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/646e224540f442589fd9ff731c3216df : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a1570f01e484f4eab639b3a455fde05 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2585336de0874403a51f710aba8f1e65 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a40af95ab9c4eef90bc947d59acb817 : SUCCESS in 4m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/017dc540ec954a659d94503c107c4d86 : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fef8face45c24c68a89a0285d58bdb6d : SUCCESS in 8m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7ce5e0b2f9ec4ec687a1b69d672380e7 : SUCCESS in 21m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1699ddbee3f744b39a74c9f524b0899d : SUCCESS in 27m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d1efab21edb04b1f8fcaacf847f7ffdb : SUCCESS in 21m 38s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/39a9154f85dd4f5285da717575a98a60 : SUCCESS in 4m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6f74a0964f6b4c30bbcf447b6279217a : SUCCESS in 6m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dd129be313624f84ba6ee55a8e1a5abd : SUCCESS in 4m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7ccca468299e402d95108ec5dee03ced : SUCCESS in 7m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7f10fbca2be54084b34a623afff18c34 : SUCCESS in 4m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/66943ee343894609970f29f7c5c85e47 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2715c3fc7b4144cf9b0671913a8f0e52 : SUCCESS in 7m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/568fa979771348e18d7bbec49f66f36d : SUCCESS in 13m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca438186197a4993a3bbd54ed9ed931a : SUCCESS in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/885dc2e36d6147f0ad1a1c22b8dc4cc5 : SUCCESS in 8m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0c9ded34c24c4c688012582496a1dfa7 : SUCCESS in 12m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/af583f7a23c54808816353c861127e99 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/15e160a49ada4234be9a4a3c8d7fe063 : SUCCESS in 8m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/06001517fc6b4285b9e7a2ccb1916b50 : SUCCESS in 4m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/03fdbf39ff7d49309bcaa526a3853d3c : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1fb881e3977d4d87b4c8758407a6ade6 : SUCCESS in 11m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c096c3764def45c692bcb5bb4f507371 : SUCCESS in 16m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/926fa01d968640e7a2efb172d2e128e8 : SUCCESS in 7m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3e4999c98d084423848b6ffbf9eaa33d : SUCCESS in 17m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/718af64159424dbdb57862f6c13295ba : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f9696970898b4c2ea3e050f16f569281 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/406feea1ac4941a4b12a0c6266951fda : SUCCESS in 24m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1eb818aea1a448638450bb1247a2b465 : SUCCESS in 7m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b587f08488b1452a9ab4d5a8dc96c816 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8bf0dcbfe9d14d29a06a945edcf87376 : SUCCESS in 6m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f1f0921a6e2f4dedba0daf137732c821 : SUCCESS in 14m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ad50d0995e114ea98d913ff04e6fddc6 : SUCCESS in 6m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b94c7d202634d0a98434df1eb5e02af : SUCCESS in 58m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d743c9e836334ffcb98ec9b3920bbded : SUCCESS in 1h 47m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9967d4bc0a9e5fc3f57e9611936a06c7ce2bf1b3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 09:56:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"515e64d0c7af091a35b38d2facee9f9f58b7db17","revisions":{"515e64d0c7af091a35b38d2facee9f9f58b7db17":{"kind":"REWORK","_number":1,"created":"2025-05-23 11:06:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/45/950745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/950745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/950745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/950745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/950745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/950745/1"}}},"commit":{"parents":[{"commit":"4a11be4ff80ae6666274712107b629a29af77e2c","subject":"DNM: split u-c bump part 3a","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a11be4ff80ae6666274712107b629a29af77e2c"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:38.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:38.000000000","tz":120},"subject":"DNM: split u-c bump part 3b","message":"DNM: split u-c bump part 3b\n\nChange-Id: I0f311f73a35ba6d7211d92997b4047aef9524942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/515e64d0c7af091a35b38d2facee9f9f58b7db17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/515e64d0c7af091a35b38d2facee9f9f58b7db17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
