)]}'
{"id":"openstack%2Frequirements~950746","triplet_id":"openstack%2Frequirements~master~I91f0db1a66f140cac5a4179ec7fa147aee7da183","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-24 08:33:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I91f0db1a66f140cac5a4179ec7fa147aee7da183","subject":"DNM: split u-c bump part 3c","status":"ABANDONED","created":"2025-05-23 11:06:58.000000000","updated":"2025-05-24 13:35:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f383f94bff6f88a4080198d56999891dd86a56f","_number":950746,"virtual_id_number":950746,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-23 11:42:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 11:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b3ee5c44b1cc8fd86658543b371dd512c08ffaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 11:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e91fb283655c535528783a017a15c08c4472f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 11:42:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6ee013bd1954aaaa1ff1e48798218ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abf6785ba1c04b7cbc782dd847dd2840 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edaedccc23db462c970ad97e9aa3bcab : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b0bd6e0c7884a24965676cadc22e892 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a996ce887e944ef2a03750fdf9e5fe9c : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/996eb365d3e84fd7a97da46defbdb86a : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf596fdf6f3f4554a94064126751e105 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0bb6aae5007a46528dffde6281364ea3 : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/401d8e52bc8e4cb19ab41d923c6a8963 : FAILURE in 5m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c761ba3cc2ea4022aa6c30ded1a6eb96 : FAILURE in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c3f165c45e944709e0ca336c194239b : FAILURE in 6m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/57f248fdb1124025880f7896c5c2e4e0 : FAILURE in 14m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/03bce4b151db487f82837af927692e36 : FAILURE in 16m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/529e497f10b64da4be8cd577f89eaf0a : SUCCESS in 6m 34s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9443a932e4524991b6910c141b1443ec : SUCCESS in 5m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/93a6de8d6ab8497fabdbce0386e2ed0c : SUCCESS in 5m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a2f093e70efc405d91a22e1f0c3cf65e : FAILURE in 4m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d47df29588904c4295f0b2779facdbf1 : SUCCESS in 6m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/027209793cc5435c9e1cd3230d97e12f : FAILURE in 3m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/df3f787255fc4246a1faec8929c7ed41 : SUCCESS in 8m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/151acff2fc4b4b89a41ebbb6e9b5367b : SUCCESS in 14m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/193d95298c294ca1adb823a385b445ac : SUCCESS in 7m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7d13cc5a47e747969b7889f85ad1e613 : SUCCESS in 8m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/029aaa63acc64081ade86677c88302b6 : SUCCESS in 12m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/94274244a3bb4bf4a7638e6486153124 : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/01e4a025da9a47c4a53d6a0e44ca9705 : SUCCESS in 11m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/69950efbb05f42a8a3ad28b02c7c57d5 : SUCCESS in 4m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2c91f864d18c4dfdbba6df34e83090bd : SUCCESS in 6m 23s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7a2e0ae297ad4bc3871d9888bc7fb0c8 : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/847981e860f74d0c9ec2830cea54cabc : SUCCESS in 26m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/eb129e9b4de44dc4b56857b1d08ea291 : SUCCESS in 6m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fdf940846d0d48ed96e7b000b5f6cb2b : SUCCESS in 13m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/29cc31495c9547fdb288ef41db926cc8 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20becdbcedd7432d805b553ba862f51a : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bad77e1606f14cd486b20cf11893f6c6 : SUCCESS in 26m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b8748f900f144eabaae2e5126a894a50 : SUCCESS in 6m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/db972c582b9f442995133d05953770e5 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e94d014fab5c44559d1207fa4abbb37b : SUCCESS in 5m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d249b85f45934a0794f2d9e76bafdef8 : FAILURE in 2m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4f289d772d464634a6860cf6ec0e380e : SUCCESS in 5m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17f46d11e5744d9c876719655a278481 : POST_FAILURE in 10m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cafe7b5dfdb9433699ccef97fee70e38 : FAILURE in 14m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6f383f94bff6f88a4080198d56999891dd86a56f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"176ce16dd756b7fb9bb4e865f273d98aecd61d41","revisions":{"176ce16dd756b7fb9bb4e865f273d98aecd61d41":{"kind":"REWORK","_number":1,"created":"2025-05-23 11:06:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/46/950746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/950746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/950746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/950746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/950746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/950746/1"}}},"commit":{"parents":[{"commit":"515e64d0c7af091a35b38d2facee9f9f58b7db17","subject":"DNM: split u-c bump part 3b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/515e64d0c7af091a35b38d2facee9f9f58b7db17"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:44.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:44.000000000","tz":120},"subject":"DNM: split u-c bump part 3c","message":"DNM: split u-c bump part 3c\n\nChange-Id: I91f0db1a66f140cac5a4179ec7fa147aee7da183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/176ce16dd756b7fb9bb4e865f273d98aecd61d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/176ce16dd756b7fb9bb4e865f273d98aecd61d41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
