)]}'
{"id":"openstack%2Frequirements~950747","triplet_id":"openstack%2Frequirements~master~I99c19e7ab04c27768d4554257fbfe77a711cb2a9","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-24 08:33:52.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I99c19e7ab04c27768d4554257fbfe77a711cb2a9","subject":"DNM: split u-c bump part 3d","status":"ABANDONED","created":"2025-05-23 11:06:58.000000000","updated":"2025-05-24 13:35:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"05fcfadac9c254527bcbc7e4bae9e3a833c0643e","_number":950747,"virtual_id_number":950747,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-23 11:37:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 11:37:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f37043e7863105c554a10b8087a4c9716734158","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-23 11:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a4c40c307c162693db9936a03b163ef4066af30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 11:37:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ef1006a6d8c4981be466f78761763e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc7663cebd14651966592aa8e16e5f7 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6aa808a83c4c25a7ea30d3574126aa : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68971b66dd0e49f3af4cdad1afadf635 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c39560c8f0d4c41869314616b581ecc : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bd12b505e5545fe8d6bfb7b82f79c2c : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fdb9302953845faa0ccc6870504fd19 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c03f459b7e7249f3aa4c4eb0a5ac5d54 : SUCCESS in 3m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dd2d4c7181a247388ccf6f4c2edcf4ce : FAILURE in 4m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c4772189221e4ba792657e048052fcd5 : FAILURE in 3m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b912e15e15944d708e00241838e393e8 : FAILURE in 6m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a9db26c484aa400d81dcc7ec32039bab : FAILURE in 14m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8b66dbb821534a70bf41d0ba5a0d6c1a : FAILURE in 13m 59s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/feacfb7d99d84f198fb7ccc8fab2d4ff : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e53fae60368f441fb7b6c036823b4d99 : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9e574e41c8f4462f83bb1290b420b600 : SUCCESS in 6m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3154fb2071ab445a9c49769201cf4f56 : FAILURE in 3m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/77371429078043779e5ac3e34b2ed26a : SUCCESS in 4m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d27e99707ef04dea9df2f1691a505d7f : FAILURE in 3m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/82868b38f6d14f038eabffb6bdd2303c : SUCCESS in 7m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8469448255f24a7299986eefdbf4bcb9 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5842dd0684b745b28c7fd76e44088e5e : SUCCESS in 10m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2bf857dafb564e01828d246d03171856 : SUCCESS in 9m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b522edf5e4e245cb9dcfc389aa0e5b25 : SUCCESS in 12m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/81c323b16d4e4f529d18b14235af62de : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/95490e1f8aa94cb5807133fab7b7f515 : SUCCESS in 11m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d9d6812327964a8688b3602fa159ad68 : SUCCESS in 4m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6446b0e0c9df46b0aff1e93c91332c46 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6b018444c69e464989cfb19ab365ca8f : SUCCESS in 11m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f66de77780434c14997d14c369b7df17 : SUCCESS in 16m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/adb7ec0afe2e4974851a0c3b045ad237 : SUCCESS in 7m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3d9d0ea84cf44b69b3d7a845ee115924 : SUCCESS in 14m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0babeed67beb465caa3ac59fc6feacc7 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/14705504cc644b10a1fde0c7b0bda6da : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9461eb0e1d1d401aa019d189ee904482 : SUCCESS in 22m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d95491dca663445985ec1df42eb1dfbf : SUCCESS in 5m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7364274d22f54391826f3ade3bae4c17 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74ed65221a9f45ecb5a9f5e5033ec98e : SUCCESS in 5m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e4e29bde50c943879a2400fe993bc2b7 : FAILURE in 2m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2163043d4a65421f8dc2919698c0ed87 : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30746d767fb54889826da1ec9c314962 : POST_FAILURE in 13m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44a463dd41d34f6eb71e5dc4cf6c846e : FAILURE in 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"05fcfadac9c254527bcbc7e4bae9e3a833c0643e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da2272471dd1694c7d9c8b3deda43c71f79e4a53","revisions":{"da2272471dd1694c7d9c8b3deda43c71f79e4a53":{"kind":"REWORK","_number":1,"created":"2025-05-23 11:06:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/47/950747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/950747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/950747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/950747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/950747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/950747/1"}}},"commit":{"parents":[{"commit":"176ce16dd756b7fb9bb4e865f273d98aecd61d41","subject":"DNM: split u-c bump part 3c","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/176ce16dd756b7fb9bb4e865f273d98aecd61d41"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:57.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-23 11:05:57.000000000","tz":120},"subject":"DNM: split u-c bump part 3d","message":"DNM: split u-c bump part 3d\n\nChange-Id: I99c19e7ab04c27768d4554257fbfe77a711cb2a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da2272471dd1694c7d9c8b3deda43c71f79e4a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da2272471dd1694c7d9c8b3deda43c71f79e4a53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
