)]}'
{"id":"openstack%2Frequirements~950780","triplet_id":"openstack%2Frequirements~master~I36c1a0a796781cdbc87a7e5f1058ae6afe14db4c","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-03 13:00:16.000000000","reason":"removed on reply"}},"hashtags":["generated-by"],"change_id":"I36c1a0a796781cdbc87a7e5f1058ae6afe14db4c","subject":"Comply with Policy for AI Generated Content","status":"MERGED","created":"2025-05-23 16:32:32.000000000","updated":"2025-07-03 14:46:07.000000000","submitted":"2025-07-03 14:45:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950780","meta_rev_id":"4c54a4d46ecf93408240e8655d3a99ccbb9925df","_number":950780,"virtual_id_number":950780,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-03 14:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-05-23 16:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-03 13:00:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 16:36:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-05-23 18:12:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"315d4cb0242af4af166051b931e04f6ba41ee699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-05-23 16:32:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c916bf86d2082563a4dab09d7c7e4525a853686","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-05-23 16:32:53.000000000","message":"Hashtag added: generated-by","accounts_in_message":[],"_revision_number":1},{"id":"387ae1544d65d5662ac0cf15b88df87fd971ea8d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-05-23 16:36:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"884040484c8bc5eb4257c83102dfd4265c92fe79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 18:12:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b86897a7f84b9796fea9229d2b04e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46272f04c2344cc3808722a1bc6a67e8 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c29dfe086003415f98ee82bad846cb6b : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e86ae34d560415ba387ac45d56ae7a0 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ddcb54b5658491caebb5407686d5f27 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cfcf9dc1f2946fc9352fe660e57e938 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2d2de0bc88d4b2cb942636b9cba6c28 : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a9dcece328c4b0895f4fd3827ef9df4 : SUCCESS in 47m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6212135b10741d6ad9bcd20000e443d : SUCCESS in 1h 35m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e9e5c1c313cb0c4c5ae19f355a6b9f93f5c2541b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-03 13:00:16.000000000","message":"Patch Set 1: Workflow+1\n\nThe similar openstack/releases change Id798bba638543cf8e9573c10e85495e7cc436237 merged, so let\u0027s go ahead and land this too for consistency.","accounts_in_message":[],"_revision_number":1},{"id":"0a854d91a35a8253836d52d2b47e150b7308e3f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:02:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27d8eb040b6375c934089187d9ff5f2021bbff95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:45:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e04c7969a944f079b4ec5429af138e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b5b890b6574e0f8d501902a269f229 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a03167b5b8d4661ab881b6ce50e6b72 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82da62c52ff842cca763b190b06db17c : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49b672a951b047f193da42536f83d6d3 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3ad4bbb00884f23964fe7280bfbdc26 : SUCCESS in 4m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63d7e449af34494787d7501d9470d84c : SUCCESS in 56m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a01b47581d7432b92acfb1ce10de957 : SUCCESS in 1h 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3844e2d520c24428d44dea9f6af07ffaceab4976","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:45:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c54a4d46ecf93408240e8655d3a99ccbb9925df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:46:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ea242d1ce942b3868d12b426f12f05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014f7fe8b9884761934be0c9073443d4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f6e4921c48c0767dd1d0b0d889734c7e66570f8","revisions":{"9f6e4921c48c0767dd1d0b0d889734c7e66570f8":{"kind":"REWORK","_number":1,"created":"2025-05-23 16:32:32.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/80/950780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/950780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/950780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/950780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/950780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/950780/1"}}},"commit":{"parents":[{"commit":"c086d80ef7ac46428281727851047410a4085e59","subject":"Make scripts DCO compliant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c086d80ef7ac46428281727851047410a4085e59"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-05-23 16:31:48.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-05-23 16:31:48.000000000","tz":0},"subject":"Comply with Policy for AI Generated Content","message":"Comply with Policy for AI Generated Content\n\nWhile probably not meeting most people\u0027s definition of artificial\nintelligence these scripts do create patches. The OpenInfra Policy\nfor AI Generated Content recommends to:\n\n\u003e Add a \"Generated-By:\" label to the commit message, and explain in\n\u003e comments or the commit message any prompts or background context\n\u003e the reviewers might need to fully understand the change and how\n\u003e much of the change was generated by the tool.\n\nEven though it also has a carve-out for these sorts of scripts:\n\n\u003e The exception to the rule is that we do allow submissions from\n\u003e well documented automated processes, such as release tooling or\n\u003e for internationalization updates.\n\nhttps://openinfra.org/legal/ai-policy\n\nNevertheless, documenting within the commit message the location of\nthe script which was used to generate that contribution is helpful\nfor a number of obvious reasons. Let\u0027s do it anyway, even though\nit\u0027s not technically required.\n\nChange-Id: I36c1a0a796781cdbc87a7e5f1058ae6afe14db4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f6e4921c48c0767dd1d0b0d889734c7e66570f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f6e4921c48c0767dd1d0b0d889734c7e66570f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
