)]}'
{"id":"openstack%2Frequirements~950811","triplet_id":"openstack%2Frequirements~master~I8e745299df55f7c63b13e3c72ecbc38641cac92e","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I8e745299df55f7c63b13e3c72ecbc38641cac92e","subject":"DNM: split u-c bump part 5a","status":"ABANDONED","created":"2025-05-24 07:45:43.000000000","updated":"2025-05-24 13:35:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fca1abc8882f80ad6419c62cf731b3fa8f5ebb31","_number":950811,"virtual_id_number":950811,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-24 09:30:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 09:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67493f23eb04198ce89d1dff5975af0b89727e91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 07:45:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd05360b454a18c4e341d0f7d26971213ed6b2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 09:30:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b9cdab4eff4e6fb247e5e872fbb9f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd6742ae4714b3684bd4a05f1f742d4 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/990d220198034e2b8029ae76f253deda : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e65b978d1f34492984865ecba4789ebd : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d3db83fd45b448e8b302f95a7026944 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46b2154bfd394caa84af081bdc9bf0c7 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d74632762cc4e3ca8d2214ddfd97504 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9ca4b7a94244b939f86f1630acd5035 : SUCCESS in 3m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/58530d699e484c9bbf67fa8574a09a61 : SUCCESS in 17m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/affa8583928e469486ad1518c42465ff : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31680153474e4231855da6e23fcadab5 : SUCCESS in 17m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/124d91608bf748d3be11c700c2599b3a : SUCCESS in 16m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2169dc99235849a69c27fb65122e7f90 : SUCCESS in 23m 45s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/68dd80ec3db34c4dbc929ca86478c12b : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e3c8ff3f81f1485f99598e6fb10e5e0f : SUCCESS in 6m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/93122000936847719463be853b1b8d3c : SUCCESS in 3m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9db603b0b461427ebc033f4a009cfc53 : SUCCESS in 11m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/63fff0a54817463eb44b1cc0f97601ef : SUCCESS in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/620b7ef77d4e48cb8137e42d1a8062f4 : SUCCESS in 5m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b5ce82370e0041c4a74972fd27857921 : SUCCESS in 8m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1e6ee1bbdce14119b024aabcf5e4afb5 : SUCCESS in 14m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf771d164b1e4a9481aeee27458fc206 : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e53713a87dc6425c8b906d132d425d2c : SUCCESS in 8m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ae7ccf75d1394f568d5a451d3207d875 : SUCCESS in 14m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/75045b77ec78414e8c586dc4e88df6e4 : SUCCESS in 2m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8f332559206149bbae9b008323b077cf : SUCCESS in 13m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/dd4bc05ab25e422f9fe43c0e97de8852 : SUCCESS in 5m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d09f224525b64633a779008be85247fc : SUCCESS in 4m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d9f3e01af078412ebbfedc7eb4ade60c : SUCCESS in 12m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ab98cb9b4aca443a930a57c2583c47aa : SUCCESS in 21m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f308e3fdaf934603970789e342d1515f : SUCCESS in 8m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4e0fff3bca0641f682d4ea64ed47a9a5 : SUCCESS in 13m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5effbf92d525438285e00c3b98f01ef1 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d2521d4ed0f4c9b9e13e26c9cbbba9c : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d5ebf5cd60b401eb46c9213fb6e03f8 : SUCCESS in 27m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d8e4d5cd629b4c05b378e54a388dcff1 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0cccf02a80c246569bafb5ebea7accf7 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a62ac0801794063a74d28fe9674043c : SUCCESS in 5m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/91ca80f696b649c180fc042727b23346 : SUCCESS in 16m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f3fbc75ef28c4bb9bf7430395cf0e901 : SUCCESS in 5m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88692786097f40368e12611a003d0398 : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38f4d6da49334b96a858134c9a256523 : SUCCESS in 1h 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fca1abc8882f80ad6419c62cf731b3fa8f5ebb31","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a05ab0e5655ab4997a5d72d23dbc9cf579f475c1","revisions":{"a05ab0e5655ab4997a5d72d23dbc9cf579f475c1":{"kind":"REWORK","_number":1,"created":"2025-05-24 07:45:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/11/950811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/950811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/950811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/950811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/950811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/950811/1"}}},"commit":{"parents":[{"commit":"9df4c3b73418755140fe442a389552bccaa6ea75","subject":"Merge \"Deprecate openstack/openstack-ansible-tests repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df4c3b73418755140fe442a389552bccaa6ea75"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 07:44:21.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 07:44:21.000000000","tz":120},"subject":"DNM: split u-c bump part 5a","message":"DNM: split u-c bump part 5a\n\nChange-Id: I8e745299df55f7c63b13e3c72ecbc38641cac92e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a05ab0e5655ab4997a5d72d23dbc9cf579f475c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a05ab0e5655ab4997a5d72d23dbc9cf579f475c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
