)]}'
{"id":"openstack%2Frequirements~950812","triplet_id":"openstack%2Frequirements~master~Ie9c2fd19390daf63078ee92bb877a8442875bc49","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-24 13:35:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie9c2fd19390daf63078ee92bb877a8442875bc49","subject":"DNM: split u-c bump part 5b","status":"ABANDONED","created":"2025-05-24 07:45:43.000000000","updated":"2025-05-24 13:35:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"684467baf52034b866d4be80c6243b48d9c23aed","_number":950812,"virtual_id_number":950812,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-24 09:23:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 09:23:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0d914957e706b84bd093ec21d427241fb5638a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 07:45:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"779ef3ef4823f43ee8814eafcfdac0f79c8d52e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 09:23:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de456e6c14a644c8a51cd3aba5c7941f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac79127138c544059a7cc5606f346dc8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7d39e65a9e40fb8a2c0f0ad002a6bb : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b92147f1e0a7449392f2b8d201554fc7 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb9f78ba0bf5433c94f572dd443fda80 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f23fbca7956a446a828f95f542b07ffc : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a41c81a9b394292a96f5ab919a23260 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1fc8e67a1c247a1a474471fc656dd13 : SUCCESS in 3m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6669d72b223347be8db09c13c8dadeb4 : FAILURE in 18m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/51dc2375d75941279bcfabc6865d3b6b : FAILURE in 9m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bb79dd9670604ebdbe073ff7436c842c : FAILURE in 15m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/56f8cb630e12424083c5635640755423 : FAILURE in 16m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/020ea6770b2b4461baa6ed2fd6410812 : FAILURE in 19m 53s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9821bcd3880f48c5a1ae47d5f67e9232 : SUCCESS in 4m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7cc41af479d74ceb8963d5c3b4213873 : SUCCESS in 6m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6ae24b09e5404ce090e86ec3ed1cfb50 : SUCCESS in 5m 23s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f80da7274488438d9db2a0fbb5589446 : SUCCESS in 11m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/09e18349b12747809718361cc34f9556 : SUCCESS in 6m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b044368298dc4db6bfaa95b00a23bf65 : SUCCESS in 2m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4bbdda4dbe30430cbd3fb4e659dcf494 : SUCCESS in 6m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b24a9e82686c4944bfc99defe8d94b1b : SUCCESS in 14m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b52fe00d820b4331b735fc175dcf2489 : SUCCESS in 8m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b45fd4d93ca148d69119e529fa0063e6 : SUCCESS in 8m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4228ce6be7804c80bc8804daa17d5e12 : SUCCESS in 11m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6b2969fef74f41d4bbbcc43c81759744 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1d0f6f45bb044da69a09eef4fa2534ca : SUCCESS in 10m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/68a75e73a80f43c5aa4891b6112dd56b : SUCCESS in 5m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/274b5f7e8e704eb989dc29b1ec00f865 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/aebddfd3a1514e1d9e617015832da6cf : SUCCESS in 10m 50s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0f5292e407f843f7ac0604c484b7d7b2 : SUCCESS in 18m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/34b3f84d7cbd4107a77707a69534663a : SUCCESS in 8m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a000e80ad85643aab0df35d791bc001b : SUCCESS in 12m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0dc541a309954e85b8f10ef496510663 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2b9f4812dee40fc82e29c69c1d8afba : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1860c65c309c4092bc62cad758cd98e1 : SUCCESS in 26m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/aac6c8ee82934bf5abe40287de1f76a4 : SUCCESS in 6m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/98eae26cfb46427cbc09592e5e308ba8 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ec2c67c8e7d4f408390af12d2c0a88c : SUCCESS in 6m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c3989fc3dfbf428ea5b28d07f23c9ece : SUCCESS in 17m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/baa563e8064a427da202bd2592695063 : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff8c5f38c78548ae9d3312c7ff2c1c7b : SUCCESS in 57m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf8d46925e8e4052b4931f7bb5751cb2 : SUCCESS in 1h 25m 39s","accounts_in_message":[],"_revision_number":1},{"id":"684467baf52034b866d4be80c6243b48d9c23aed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf017f94ec5040336e368cda961fc9ad745bb6cf","revisions":{"cf017f94ec5040336e368cda961fc9ad745bb6cf":{"kind":"REWORK","_number":1,"created":"2025-05-24 07:45:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/12/950812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/950812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/950812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/950812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/950812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/950812/1"}}},"commit":{"parents":[{"commit":"a05ab0e5655ab4997a5d72d23dbc9cf579f475c1","subject":"DNM: split u-c bump part 5a","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a05ab0e5655ab4997a5d72d23dbc9cf579f475c1"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 07:44:38.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 07:44:38.000000000","tz":120},"subject":"DNM: split u-c bump part 5b","message":"DNM: split u-c bump part 5b\n\nChange-Id: Ie9c2fd19390daf63078ee92bb877a8442875bc49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf017f94ec5040336e368cda961fc9ad745bb6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf017f94ec5040336e368cda961fc9ad745bb6cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
