)]}'
{"id":"openstack%2Frequirements~950815","triplet_id":"openstack%2Frequirements~master~Ibd81e50f6b552d8ee6f2b3aa79bb3d91a2e56aed","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-27 08:38:03.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ibd81e50f6b552d8ee6f2b3aa79bb3d91a2e56aed","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-05-24 08:25:28.000000000","updated":"2025-05-27 10:03:51.000000000","submitted":"2025-05-27 10:03:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950815-openstack/requirements/constraints","meta_rev_id":"dba5a102749dbf4ba0c9948c858b82fe22d1fd01","_number":950815,"virtual_id_number":950815,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 10:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-27 08:38:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-27 08:38:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 10:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-27 08:38:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d4ccf8ca4437ac71f34661145a1a2647cd0c63bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-24 08:25:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5b09a5f5cb9548f9f6f1c76255aa864b977c8aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 10:08:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cc24bce9b7c4cb0abaf5698261df876\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171fb94ce51e41f7973cfde31c1af4cf : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a86ae78d374174928cc5cae369c47d : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/510882219a28490aacca76b986ee8874 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fde0b73145040269132e2cc91bce9f3 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9ceda10b65c475388aec975ea2e88b5 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2886cb7e6f2427381c55f14020dad74 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96c24a65d0cd40d3a1667375953d8487 : SUCCESS in 3m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f959341c260f4f2ebbc0f02cda394300 : FAILURE in 2m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4c4be2651eb2484da5d492c3771fb0b1 : FAILURE in 9m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5fcd26eb3cdd45c1a6be9a21ab4ebf75 : FAILURE in 7m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/671036a4efc54264bab66228df6fa7cb : FAILURE in 14m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0835f45191604d8599d0e3fa410e79b4 : FAILURE in 22m 31s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0d988d466b2b4f448161506e6b2243d9 : SUCCESS in 7m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5159ebddf04d4d2ca58aec5b89ae2f98 : SUCCESS in 7m 02s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d43c0d45784b4e328207bca5846e1603 : SUCCESS in 3m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1dee60d6929b49188d8990dcdf32806a : SUCCESS in 9m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/90d49a7f74824811b3fbe5bd7fddbf10 : FAILURE in 6m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2540090b5d3f44b1b8f2f9839e0a36e7 : SUCCESS in 4m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fdc9a283de074cfd94000ce6ed40b1e8 : SUCCESS in 9m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9036b2d670924ee5a6efa07baa81759d : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0134ddba461c4159a2a62453275cfc06 : SUCCESS in 9m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3dcf043ada4d4d68839ffeec8becd088 : FAILURE in 4m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/073348d2772b434f9339de66b7cc4561 : SUCCESS in 7m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/eb5a106975f243afb80404ab0d296f9d : SUCCESS in 3m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c90318f38b0441b89421912a827c5c60 : SUCCESS in 10m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d9f9199bb47a4c3db9581bf94000e220 : SUCCESS in 3m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6860821b4360481d8b4e8aed322a7328 : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f8353c222d764af4b62e217eb612fb5f : SUCCESS in 11m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8e0d23d311c94954b4e5d3772c020959 : SUCCESS in 19m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9ffe0f60eb9d48ee941a9279b2184178 : SUCCESS in 8m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9cbcbfdf66164428ad82caa7e7090ee5 : SUCCESS in 13m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f407b5c3525a4e2eac15a0679e0852eb : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/172133a948264785ba669bf8120a396c : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b6d73a6f5a546e38bf6708c9bae54be : SUCCESS in 24m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b523bd180b87467d8d69d3cfafd00a88 : FAILURE in 5m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5518b66c69e940b4a0986185c8f97dee : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b7a13cc8b744cb5a18e9ced8e1cdad8 : SUCCESS in 6m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/451020aae40c4566a0760e922e797be6 : SUCCESS in 17m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/47dd7720db814e0fb19cea23f8b6d40d : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c325dfc1b954b44a0752c6b92c1d4d0 : SUCCESS in 53m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aadc3489f0234ee6be81fe154a03303f : SUCCESS in 1h 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6b0f304c832bf43abe2ba12e838ef7b0fc5bca48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:29:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ce28eaf00b45dafaa5d25d77b8fe5f5e63a1d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 15:16:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22899a004a8b4521be5afde4e14eb3e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae3b9e8dbbc436ba3317d7408bd4c9a : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21b69d944a444088a2a368e1dfdba4a : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a42b8e7bb2145d0944ca53481ed1816 : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd37cde074764faa9b3bc99a41f5d74f : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dcd4518ffa24662a992d0790f002f79 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe3cacfb7567434cb94fb901a5bb3d26 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85e324d3087e407685e4c03e2f0943cd : SUCCESS in 3m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f95f5c3375c9490e90de3839c369909a : FAILURE in 3m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/790ef840eb88418db0224632c1c5bb84 : FAILURE in 8m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6cecef0c766549d7bf10c7ad75f4acd7 : FAILURE in 12m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/55816a9100e84b2f81fbd88077faef1f : FAILURE in 17m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/165f31df8aad4508a38584b15d5dc60e : FAILURE in 19m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a31e30f665bb48dcaa7c676b21cc717a : SUCCESS in 4m 26s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fda3797838a7447a97907e8619fd5701 : SUCCESS in 4m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/65bc5599fa18457ebdf03f126957464c : SUCCESS in 3m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/edee93dd577e457cbf31be10347a704f : SUCCESS in 9m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b448d7b098184b28a51cda2184adf1e1 : FAILURE in 4m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f46c57bc3964425c8d6f70e446fe1f6f : SUCCESS in 4m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e7981ec7ec34480fb4ca35db8c535655 : SUCCESS in 6m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2c028c3ee72041848d348cd77f112bea : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0461544036d94422a12bbd61be6c845b : SUCCESS in 5m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/97db548b2d7045e7abbee032844a9a2f : FAILURE in 4m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2701d33a4e1349faaffbd782b6361003 : SUCCESS in 11m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/235845c21d854b5fa97bff7215094c31 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f7c3f7e26b3548e3b5c845ca037d0a70 : SUCCESS in 9m 56s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ac09c87983fe4375be9e2b9e2d951a32 : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/09ab0b5276db48c58a9720c74c589884 : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bba8d4a22df849b3bd804e4f268cc435 : SUCCESS in 11m 15s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ecaa238dd5a04283bb4db09b6854565b : SUCCESS in 13m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f3c28daa105648a98f441361f9a2bfb4 : SUCCESS in 7m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d7d406c8de0b4e3c853d999f9a8402da : SUCCESS in 10m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fdafb6de9c8944f1a47dfacf2362d291 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b300443ed7b64ef48483e73c207e7809 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f12b4710e3540e586cdcf71a778381a : SUCCESS in 19m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/40dd5f289efc42148a88edb248956fb9 : FAILURE in 4m 36s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/05d427bc35864a83ad8df21747882b53 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/538f46005f004d42b6087dad1aabfbed : SUCCESS in 2m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/66180d359b9c4e5bbdb736117d67699d : SUCCESS in 17m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e9bcc24e97344f75aab4d2cd5e8d858c : SUCCESS in 7m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8cfba97d7fb4d05aae5255b24abe6d8 : SUCCESS in 57m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a54c9dfe70e4e0fb9fc602afc81e86a : SUCCESS in 1h 41m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4e68891c1aa78a9b7f23db54627880691d6f14a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-25 12:26:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b98a277550965a9cbb53297ce0b83172bdac37e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-25 13:48:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8c9e04a6553497590c50635b201b851\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34741f0a4a054df898ac00430087a112 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c48c3c1f06456da7e18dff19dac03d : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41211ca9bb554a839a17f8cf48a87727 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a48208c9ad4d42c99352774f4b90dc25 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eec31f8ff4f944f696ca6c236214002f : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01b5c044fe1243baa168faac23d0efb8 : SUCCESS in 1m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4d6be459b044c9fae3a9db7d7044d4b : SUCCESS in 2m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d305f38eadde435ea9d0e1adca43bfc2 : SUCCESS in 17m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2d7ec4d5e0224c028e97375848b15fdd : SUCCESS in 9m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3b425853492d4811974fb9e04edac399 : SUCCESS in 12m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5601d29290174b1cacc37a7b78df5abd : SUCCESS in 19m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/45aa6f41f1ab4d6dbcee5618c78045d1 : SUCCESS in 16m 20s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3ebf575e05094346bf5ce2896a755ebb : SUCCESS in 3m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/113ccccad15d4dc88ff4317be25a42b5 : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e39077254464452ba39ab1aafc53c82e : SUCCESS in 3m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/87ff0fb6968442b7acb40b7c1a17b77b : SUCCESS in 9m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9dec2d55cdf84f858da93f5d11b24177 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9ebab66050b84912ba1fcf9b60098d1d : SUCCESS in 4m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/84d4f2e93ff44a0a9279b12d843bece0 : SUCCESS in 5m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6ce6b635b37041dabb0592ab2eb9f360 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c207b9b20ee146beb2a5cfc66f9ad32f : SUCCESS in 5m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/de0b88d0299748d18bd624148677560c : SUCCESS in 6m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/21f4c82f38104421afb4657edf570291 : SUCCESS in 10m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0484ff77f5614e5498ff63dc77fc3b4b : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/997505effcf245818b5843cfce769e7a : SUCCESS in 9m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a8f79a8224d64533a225678a6d8313f1 : SUCCESS in 4m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0d509e23f192489d8fda5ac8afc3f5a8 : SUCCESS in 4m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ff14e5ee7b0348aabb0c811ff72d1c96 : SUCCESS in 12m 35s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9f887a594e9d4804afe531e2711eb402 : SUCCESS in 17m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98b011c8280e404986e5d6fdb67d12ac : SUCCESS in 6m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/781f9613907f408b926e2bced0d032fd : SUCCESS in 12m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f80d21c352664216acdc7064de8e999f : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/94e970e46bac4e4993aee43ab7ade44d : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0edff41c8d984a4eb7d2147b20db8fa1 : SUCCESS in 20m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/20276d43166d4d1c8b6583dbc7621fd9 : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/867099fc2dbf44c9b9005d1c5f373800 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1421fe55fee447dacd1630c9999c0ce : SUCCESS in 5m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/13356dd9d648403baaa6a9e320b4eef4 : SUCCESS in 11m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/738deebfb0434d3f9b4bdf1d3e0b664e : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4782c3b384784f71be8776d0211e729f : SUCCESS in 47m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04ed256b69f341c0951dbf2a4e9006dc : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":3},{"id":"545d9dbe79c4501d6128b2d7cf2639530c0ef2ae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 08:38:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e74b2ddf6b0483200623113b694ebff19152388c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 08:38:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ba19fc02d4d76ef7b777e926ba85ad0ee38f085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 10:03:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e0ff118e99d4ff1ac8b723d66bc5229\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c8f829dc0c4b76b6d139b70d953bb1 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61533e75f1d46f791544c836c834039 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5020c6e11634794a925eb56eeec3167 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d36027a71af94a60857ab3199cb8980d : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5b34548cd6b498f8c7c877e0ba2f7c4 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b295235701b455c875054f9a9e598a0 : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b4baad887b744c3982d8af2458d0c1ed : SUCCESS in 15m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/59771e8f1bae4d3199d071e38ca9713b : SUCCESS in 9m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b80c8a2dd9974c9e9da74b60f9a6b2b0 : SUCCESS in 23m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f4f970aa8b1d423a8c41067a1ead3516 : SUCCESS in 18m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/203db9e5da294ba5b77c3ac1190f460e : SUCCESS in 13m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2c80de66fc9a4f7db961eb38ae835b4e : SUCCESS in 4m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/26a69393f0584145bee98127228508b9 : SUCCESS in 6m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/253eaae1bba74ac2808d5cbcfdb36111 : SUCCESS in 7m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0b591f8fe31549e99fd69a1d2caca510 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd874fea0d594919a7819c7c93da93f1 : SUCCESS in 7m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/845073c32b5148e6b14a105ccfc81ef7 : SUCCESS in 7m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aae59db88583440f9d79b9a8d38ae5da : SUCCESS in 14m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/05631fd4f1bd40d78df797ce3dcacc0a : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b3bb7eab3ff840da8db6db9a384d5e7e : SUCCESS in 11m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6427cbd296f047faa32f6bd4bf2525eb : SUCCESS in 17m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cb43a9844eca4c41ba6871ee76d882cc : SUCCESS in 7m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8a8439892ae84c9b940bb7bf8e444051 : SUCCESS in 11m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/de7b4f6dd0b34c4d8800007907acf217 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9d7ab3f78bf4aadada43dd1fbc67a65 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a58876042dd142ffb94695bb6d1c6488 : SUCCESS in 25m 38s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e7f3b51245094e2f938c3392dd5f8ed9 : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/046fd10b1be2414e8bf88fd3d7daecf2 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f4b40ae76824e2d9d5ff3cafe2a544c : SUCCESS in 6m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/67597fdf6e344d0f9efd4c48d98a1082 : SUCCESS in 17m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/663c4cdf504a4b649217c6967357eef3 : SUCCESS in 5m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2246ea6e8bc491a8acbd05282e9ebda : SUCCESS in 58m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d75f4ff47834163a1a9f4a8a4c2bdad : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bf1fcf65d9c46ea4d2033cd80db2ebd13d909c04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 10:03:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dba5a102749dbf4ba0c9948c858b82fe22d1fd01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 10:03:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf0d00acf48418384f4aa67db15a82b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa9d8be00a94af296be5875a0b61276 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ec76c273008e3317426231cdad782622aec6e265","revisions":{"5cf616d789076c9af62f39d3323623896509d3cb":{"kind":"REWORK","_number":1,"created":"2025-05-24 08:25:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/950815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/950815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/950815/1"}}},"commit":{"parents":[{"commit":"9df4c3b73418755140fe442a389552bccaa6ea75","subject":"Merge \"Deprecate openstack/openstack-ansible-tests repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df4c3b73418755140fe442a389552bccaa6ea75"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-24 08:25:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-24 08:25:28.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibd81e50f6b552d8ee6f2b3aa79bb3d91a2e56aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cf616d789076c9af62f39d3323623896509d3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cf616d789076c9af62f39d3323623896509d3cb"}]},"branch":"refs/heads/master"},"1f81a38877e9920935067a5164d6f8fabbbd96a3":{"kind":"REWORK","_number":2,"created":"2025-05-24 13:29:45.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/15/950815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/950815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/950815/2"}}},"commit":{"parents":[{"commit":"0697f59bec1402466cfc71e33529293ac1739e72","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0697f59bec1402466cfc71e33529293ac1739e72"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-24 08:25:28.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 13:11:16.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibd81e50f6b552d8ee6f2b3aa79bb3d91a2e56aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f81a38877e9920935067a5164d6f8fabbbd96a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f81a38877e9920935067a5164d6f8fabbbd96a3"}]},"branch":"refs/heads/master"},"ec76c273008e3317426231cdad782622aec6e265":{"kind":"REWORK","_number":3,"created":"2025-05-25 12:26:56.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/15/950815/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/950815/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/950815/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/950815/3"}}},"commit":{"parents":[{"commit":"0697f59bec1402466cfc71e33529293ac1739e72","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0697f59bec1402466cfc71e33529293ac1739e72"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-24 08:25:28.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-25 12:24:21.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes for automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- importlib-metadata\n- click\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: Ibd81e50f6b552d8ee6f2b3aa79bb3d91a2e56aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec76c273008e3317426231cdad782622aec6e265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec76c273008e3317426231cdad782622aec6e265"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
