)]}'
{"id":"openstack%2Frequirements~950816","triplet_id":"openstack%2Frequirements~master~I284656c3c04b084d32f17e41d083e8cf469b14bd","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I284656c3c04b084d32f17e41d083e8cf469b14bd","subject":"DNM: split u-c bump part 5ba","status":"ABANDONED","created":"2025-05-24 09:15:27.000000000","updated":"2025-05-24 13:35:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f4e6632787b6836b352de9ef46206c4653bc896","_number":950816,"virtual_id_number":950816,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-24 10:37:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 10:37:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10837d17436ebf5dc62f1e8eb48d13482bd44e1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 09:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c22d034ce84a93f41b1fc2f4d62d1074f2e14bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 10:37:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cba8273f45074780ae85f20276ae1acc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad1dc732fc643b2923cb7d73a54283a : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0034f58676348adb3c26afdae064570 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2036ed80408642aab41297c7213176e9 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/602bc7fbb56743648622bf40623ca13c : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb9239851ead47bc84201b32ddfd65b6 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3d9f94e2db44dc897da872d319802b3 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ff1eb067d714cb78b5eaf2507cb08ed : SUCCESS in 2m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fd2cb4ffeddb474aaf8e1737570a70df : SUCCESS in 18m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/11bcd263bb6f43fd98e002f30b5fa022 : SUCCESS in 8m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b61551d920e944f89436ca7eb09de5ef : SUCCESS in 18m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5cc4bc2eb3bf459badc30e7d434e9ba4 : SUCCESS in 24m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/86162788f4bb4844ab4732a35054c5c2 : SUCCESS in 17m 56s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/078070b506664e2f9ad06d30f87238d1 : SUCCESS in 5m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4a9f20a55e424352b734fe701d6d5408 : SUCCESS in 7m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8752920901924bd38cbd046cea61359e : SUCCESS in 3m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/205cf61a3a4746659f1d55c11833a7f4 : SUCCESS in 12m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fd6c939de48543608d5e3861f84ee652 : SUCCESS in 7m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/92c0002458d3498380784c540fafa3d4 : SUCCESS in 6m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/50fbf74ebec04ec08ef832a40e92420f : SUCCESS in 9m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3a3ee255a72c47b58e73b388ed238672 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5c94d1bcce343978a65027b1958285e : SUCCESS in 9m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e4f3d7ce6910434d9abdcbd7a4e897e1 : SUCCESS in 8m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bae610434c0b4c69b6100b8614ee46d8 : SUCCESS in 11m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cb8d9706c8ba4ee1bd8e4353580e90a9 : SUCCESS in 2m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c8310e5e8bb84fb1894fb93f33d4a804 : SUCCESS in 9m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e8113eb709f34a86a14d236a28a9976b : SUCCESS in 7m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cca32996872e4f4f88cc67524d671cb6 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/625e4bc72661455ab87fa7b23e20418f : SUCCESS in 12m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/dac30589b0164428b08007a79cb3bca4 : SUCCESS in 21m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e35894b7518432e9c304816c19c376c : SUCCESS in 7m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/82357161278448c79c4c893de59c9af0 : SUCCESS in 14m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1c0087115a654a84aa461f3680bf712a : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a727194006c0409abe786f525d160de1 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/baff3fc147e146d9971354cd65e23f78 : SUCCESS in 22m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b3768a3f29394b16adb5f802c67d701b : SUCCESS in 5m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d50368da38b04d4299b4cc1766d296d4 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42b83df8306f445eab1da4b1743e5886 : SUCCESS in 5m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9b14164a759d4194a8376a694d3a5a22 : SUCCESS in 18m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/63a35414edc94495b709ba0c03f50ac9 : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84a4e263645442b1bdba88f706e5f061 : SUCCESS in 1h 03m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eaa4de9ea634441ea69a486eac22c276 : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7f4e6632787b6836b352de9ef46206c4653bc896","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e677aa1398f9f9429359bc51bf630784d751b9d8","revisions":{"e677aa1398f9f9429359bc51bf630784d751b9d8":{"kind":"REWORK","_number":1,"created":"2025-05-24 09:15:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/16/950816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/950816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/950816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/950816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/950816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/950816/1"}}},"commit":{"parents":[{"commit":"9df4c3b73418755140fe442a389552bccaa6ea75","subject":"Merge \"Deprecate openstack/openstack-ansible-tests repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df4c3b73418755140fe442a389552bccaa6ea75"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:16.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:16.000000000","tz":120},"subject":"DNM: split u-c bump part 5ba","message":"DNM: split u-c bump part 5ba\n\nChange-Id: I284656c3c04b084d32f17e41d083e8cf469b14bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e677aa1398f9f9429359bc51bf630784d751b9d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e677aa1398f9f9429359bc51bf630784d751b9d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
