)]}'
{"id":"openstack%2Frequirements~950817","triplet_id":"openstack%2Frequirements~master~I9a15ce1a29f360bb3d6d384a281aafa9d6037005","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I9a15ce1a29f360bb3d6d384a281aafa9d6037005","subject":"DNM: split u-c bump part 5bb","status":"ABANDONED","created":"2025-05-24 09:15:27.000000000","updated":"2025-05-24 13:35:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8bd8af04836184701fd4848a070633c96f901918","_number":950817,"virtual_id_number":950817,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-24 11:12:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 11:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0675d9551690b45fe336687437208a036a5d750a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 09:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef21fda62ecc002761f6144fbb3468f5004c6be4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 11:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fecbc45b0cf949ab967cf0a8a606d7bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c376c87119484f5193c7123d40ba3ff1 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391d59adc1204ef08c93a90f240b6dc1 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a698ba90f7924d99837cbb439c08c2cc : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3831891f7b43489bbd3703990d417e37 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d7f8985e07347bdb9da1e2f8e26107d : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf49c950d59245598f58d28c1dd929ca : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2fa8f815ab764f85be4860b0bb50631b : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/52ebd97238544b16b695be48b00a156e : SUCCESS in 16m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7e03934094154f108558ae3d0b26a764 : SUCCESS in 9m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/43abf03d02ed4f71aae7ac4db61f2eaa : SUCCESS in 17m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2f62daef998b4a1fbaa8dc711cec0331 : SUCCESS in 15m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b302807969d64b90a4c9c60f8087af5a : SUCCESS in 23m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9cf099b98fad41c0b9c8ce8d06acd277 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/54f6e56b397a4551a8e53ce88d7b8c29 : SUCCESS in 5m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/393d5c3fc3994bfead2f2e4ea9676b1b : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1fda204f78a1426c8b767aebf799e378 : SUCCESS in 9m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3ef76a92ffa040ffa36a167dd313d7b8 : SUCCESS in 4m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9b763e94b75843869ee50b35d09976c9 : SUCCESS in 4m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8adc9a448db44387a8e956a80a1cec70 : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5e8f61b2c24640ca82e1e8faccb9ab9d : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbb4d4569ad14593807b3ce47e9c7b2e : SUCCESS in 8m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9495f1df9ed14749bc1e8214f3b5b768 : SUCCESS in 8m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ba78cfc48330491c9854b8ab25bb1b63 : SUCCESS in 14m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a90b2e42bd7d46f1a7f4f41c2a087c30 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5df4ff94e13c4134b46f8403df98e30d : SUCCESS in 10m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ee8944fcb59041528918253da8d4ac6c : SUCCESS in 3m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0ba68194fe054bb68ffba70a3ea09138 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1c804dfe4da04becbca4a421dc2ac2fa : SUCCESS in 9m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cd567fe58c414a4fb2b562b193f2783c : SUCCESS in 14m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ef664d936fa243c6aae9a7350181f5bd : SUCCESS in 11m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2e83a0e71ba6473db10f4128c2185680 : SUCCESS in 11m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f2390e90181a4462acea78e2fc8e4aa8 : SUCCESS in 3m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e87d87e45a64c8dae618c5b68e57f3b : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ccc4ebb5136346e299f265f6ac2c5cb0 : SUCCESS in 24m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ab6c2de48b37495d81b75c438521b82b : SUCCESS in 5m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/250c8b78459c448485495b978c564c6a : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f743fe4af4b9471fae56ddb45f10911e : SUCCESS in 7m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/13e8cb4fc8594b02937041d3920352ab : SUCCESS in 14m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0ebc93be1c5049efa07b1914b13b1138 : SUCCESS in 6m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6881a0017a574747b2809c6143c800c8 : SUCCESS in 55m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/779e7779f4954b59be3a23b910caf746 : SUCCESS in 1h 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8bd8af04836184701fd4848a070633c96f901918","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e37e44c37c13aa91c2659ab30600030a1cfbc7b","revisions":{"0e37e44c37c13aa91c2659ab30600030a1cfbc7b":{"kind":"REWORK","_number":1,"created":"2025-05-24 09:15:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/17/950817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/950817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/950817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/950817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/950817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/950817/1"}}},"commit":{"parents":[{"commit":"e677aa1398f9f9429359bc51bf630784d751b9d8","subject":"DNM: split u-c bump part 5ba","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e677aa1398f9f9429359bc51bf630784d751b9d8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:35.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:35.000000000","tz":120},"subject":"DNM: split u-c bump part 5bb","message":"DNM: split u-c bump part 5bb\n\nChange-Id: I9a15ce1a29f360bb3d6d384a281aafa9d6037005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e37e44c37c13aa91c2659ab30600030a1cfbc7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e37e44c37c13aa91c2659ab30600030a1cfbc7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
