)]}'
{"id":"openstack%2Frequirements~950818","triplet_id":"openstack%2Frequirements~master~Ifd102a6f41b2ec3ab839e86380a81495cb2f128e","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ifd102a6f41b2ec3ab839e86380a81495cb2f128e","subject":"DNM: split u-c bump part 5bc","status":"ABANDONED","created":"2025-05-24 09:15:27.000000000","updated":"2025-05-24 13:35:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"759676c9f6660278a28ce08b0410fd3b8f60d424","_number":950818,"virtual_id_number":950818,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-24 10:59:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 10:59:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5518b4146f72ba7df738d877154e85d013ff8521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 09:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edadf0d718330e39d03478bc7c1569c444785dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 10:59:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8e2d351f3e4aa88c40bbb6bf191d9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f562847dfc1e4ed3bd1c1b1cd2c9914e : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc46b7714438457da179b13801f3294e : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ff9c1baf3b242b5861834b2531e3bcd : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a97ed145da6f446c8fa277efb353b445 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a589d132f0a4296abb346c584eb8654 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1037f21cc0b40a2a201d6fbbca18166 : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9734cf98093c41f5b87ff56fb47c5d98 : SUCCESS in 3m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69709b74bbda48d897b16e94d9b85dd6 : SUCCESS in 13m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/df22df973db14e10a823e0a467abf3b4 : SUCCESS in 4m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f5f03ed9f42e4b05ad04e4c911fa7b45 : SUCCESS in 17m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/309102a67ef141318c03e8d48752bbb4 : SUCCESS in 22m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3d11d305b28f428a8b0da02d19ef6f03 : SUCCESS in 28m 17s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ab7e8717268a41a49167c2d001e48b8d : SUCCESS in 5m 12s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/76577bc086a042769a441c2039cce0ae : SUCCESS in 5m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0b47e155a4514e3d9a9ab29e7d539346 : SUCCESS in 2m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7eb12c2208174215afd0883a460351b6 : SUCCESS in 12m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/44d5329db3754e508de88794075c27ac : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1f8bffbf887640f79ac4c41ce88b5e03 : SUCCESS in 5m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dec0a684188645209ccfec6d034558a6 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/18615cb7994d495f8138f1ddd8025c43 : SUCCESS in 15m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45b69bebcd0743fa8c2504bca7490fe0 : SUCCESS in 6m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/807d43e2897f4a5caf665c56fb13ca93 : SUCCESS in 8m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6ed6d4bc43294a589805080c0f355184 : SUCCESS in 11m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/71ef2e9f62f54d9da8dddf3e5b8e9a4d : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/19190b6d8ab34b4da70625a620bc93e3 : SUCCESS in 9m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4028af0426e940e9b3610e97e7d3054d : SUCCESS in 6m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5c94910633354bc0b02a69b7d228fa0d : SUCCESS in 5m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/24c3228b334649f69f85335ed5da8342 : SUCCESS in 9m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/876de03ea6be4320b593022cc5966c0a : SUCCESS in 17m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cfd1e141deff4eadaf38bc178f256e99 : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c6de9e4099304b799fbc54093232714c : SUCCESS in 12m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0dc50e39e1a347b0a005d527d5704e56 : SUCCESS in 7m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7993e7ae63a9405aab63f77279651a59 : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/648b7601fd924f03a9fdd908eaa178d6 : SUCCESS in 22m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/43770bd7bff74265b87d96e9328e4f9b : SUCCESS in 6m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/eef7e8c3c1a94a82b3834d774bc1770c : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9dab7f73c69341e3a5c1d86452096119 : SUCCESS in 5m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bb7bd0d0d635425cb69511c41eea7172 : SUCCESS in 11m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3556e70c442d47ae994b8db806402db6 : SUCCESS in 5m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bfb3ddfd9604ccf84e69584a7c9530a : SUCCESS in 1h 00m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3885ac56646a447eb8e27aa418102612 : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":1},{"id":"759676c9f6660278a28ce08b0410fd3b8f60d424","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed7f4e4f3603906bbfdbecaa152976c7d86a2cc5","revisions":{"ed7f4e4f3603906bbfdbecaa152976c7d86a2cc5":{"kind":"REWORK","_number":1,"created":"2025-05-24 09:15:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/18/950818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/950818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/950818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/950818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/950818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/950818/1"}}},"commit":{"parents":[{"commit":"0e37e44c37c13aa91c2659ab30600030a1cfbc7b","subject":"DNM: split u-c bump part 5bb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e37e44c37c13aa91c2659ab30600030a1cfbc7b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:48.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:48.000000000","tz":120},"subject":"DNM: split u-c bump part 5bc","message":"DNM: split u-c bump part 5bc\n\nChange-Id: Ifd102a6f41b2ec3ab839e86380a81495cb2f128e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed7f4e4f3603906bbfdbecaa152976c7d86a2cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed7f4e4f3603906bbfdbecaa152976c7d86a2cc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
