)]}'
{"id":"openstack%2Frequirements~950819","triplet_id":"openstack%2Frequirements~master~Iec6d9ad9e77cae41d769e0dcedf8fbc68cfb91a9","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-24 13:35:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iec6d9ad9e77cae41d769e0dcedf8fbc68cfb91a9","subject":"DNM: split u-c bump part 5bd","status":"ABANDONED","created":"2025-05-24 09:15:27.000000000","updated":"2025-05-24 13:35:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6b2bcd0dda85df05718192ee8c1338d301dcc7e","_number":950819,"virtual_id_number":950819,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 10:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f1d9e00ebb4085a6639cf6a24e99d31d4011c69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 09:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f99e3ef12013581dd8f2c732e3e50d1e6e8cee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 10:45:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb27c5ee8978442983a778e7c1206fee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c8b810379f4262a44d58265b020950 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1fc5b6faf14e1bb4e1262e843994df : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f74448463d364b5e8268e22a9b5f7bb4 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c18ec21f685473db605864fad57cf57 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21d86f51f96e44de9816dc14d74e0d3b : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/698d72612ce94b38ac0842d1c458a9f8 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e23994538c24d9e84e859a05dbc6e6c : SUCCESS in 4m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/343fcaeb3f844f719f639db8c0cd1c8a : FAILURE in 18m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/70fd319481a54807b9185deff15240d9 : FAILURE in 12m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/55eb46de2e4144dfa3371c4776fbea55 : FAILURE in 19m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4decb91708084e1fa9ab88197321bf69 : FAILURE in 14m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/680f49dd3ad6438fa698a07ee4eaf334 : FAILURE in 25m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/da4018ce09d440f2a01ce17d28a7eac5 : SUCCESS in 7m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a6ad8e226128499a98f30227dda50fe6 : SUCCESS in 4m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4503beeb4efc452da8b9e1a0dea5b0cc : SUCCESS in 5m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1c25d28d112c4c4d8a32453bcb126cf7 : SUCCESS in 14m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7c71241d8b0e4f15ba4b56e448315af6 : SUCCESS in 6m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3b5196ded0384818b5f8489820ae95ca : SUCCESS in 4m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1640c2e4af8c4c849306d41165c4c124 : SUCCESS in 6m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/74b84584eaa244809413c50e9069247c : SUCCESS in 14m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f8e46a3a29d451bb86403ab07b370a7 : SUCCESS in 6m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/effbb339cdf54c1683a4d4bfa0ac7bd1 : SUCCESS in 8m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/00591782ada148e9b65f04c29881bc94 : SUCCESS in 14m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5a2a9833f4274b23bb95530da2aa806a : SUCCESS in 2m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f8b870fa25b24794b23010fc6e13174b : SUCCESS in 12m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/eb03099276ee4e9ba34fd0e372373e26 : SUCCESS in 4m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/780560056afd4c6489aeee27a082e89d : SUCCESS in 6m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fbcee826becd4c9fa60f7c99a54b16ec : SUCCESS in 11m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ae745f8d43ff4c9face3bc279731e4ac : SUCCESS in 19m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/068c92f79a1649f08e0211557ef23ab2 : SUCCESS in 3m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/15538b3731c84fd2b764bf3353c6b30e : SUCCESS in 13m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9621a0b3ea85493cbd9a930b7bc5abc5 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6eaeb6d41e834b3bad76de815b3c2378 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/674e14710b0e46cb97b1f8c0d344a40b : SUCCESS in 28m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/41c02d2475bb4c31b14ae36a4e89634c : SUCCESS in 5m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e674b2c14b7f45dca4f4adcc2fdb77ab : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1f51c25086f4503a262f44680749650 : SUCCESS in 8m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bc430f242fbf4066b0344b2221c90a33 : SUCCESS in 11m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c454e4883cbb4e9786c9ee8955473918 : SUCCESS in 6m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6f1bb13187041dd8a7a11dcade506b7 : SUCCESS in 1h 05m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de9f734878a84f0da6f0634c1167d0ec : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d8bdc1c2f208f214166908fb83e8adc0e1c91f5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 12:02:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c6b2bcd0dda85df05718192ee8c1338d301dcc7e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 13:35:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"025f313674fe1b578ad6cd77052d7f9ee3a00591","revisions":{"bd762ab046a838fc4fee071cd5ef806e4db39970":{"kind":"REWORK","_number":1,"created":"2025-05-24 09:15:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/19/950819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/950819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/950819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/950819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/950819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/950819/1"}}},"commit":{"parents":[{"commit":"ed7f4e4f3603906bbfdbecaa152976c7d86a2cc5","subject":"DNM: split u-c bump part 5bc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed7f4e4f3603906bbfdbecaa152976c7d86a2cc5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:55.000000000","tz":120},"subject":"DNM: split u-c bump part 5bd","message":"DNM: split u-c bump part 5bd\n\nChange-Id: Iec6d9ad9e77cae41d769e0dcedf8fbc68cfb91a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd762ab046a838fc4fee071cd5ef806e4db39970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd762ab046a838fc4fee071cd5ef806e4db39970"}]},"branch":"refs/heads/master"},"025f313674fe1b578ad6cd77052d7f9ee3a00591":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-24 12:02:26.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/19/950819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/950819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/950819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/950819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/950819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/950819/2"}}},"commit":{"parents":[{"commit":"9df4c3b73418755140fe442a389552bccaa6ea75","subject":"Merge \"Deprecate openstack/openstack-ansible-tests repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df4c3b73418755140fe442a389552bccaa6ea75"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 09:14:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 12:02:26.000000000","tz":0},"subject":"DNM: split u-c bump part 5bd","message":"DNM: split u-c bump part 5bd\n\nChange-Id: Iec6d9ad9e77cae41d769e0dcedf8fbc68cfb91a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/025f313674fe1b578ad6cd77052d7f9ee3a00591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/025f313674fe1b578ad6cd77052d7f9ee3a00591"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
