)]}'
{"id":"openstack%2Frequirements~950823","triplet_id":"openstack%2Frequirements~master~I906c817f144d4a1d59c1385b7f910c685690377c","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2025-07-15 15:24:55.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-15 13:48:16.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I906c817f144d4a1d59c1385b7f910c685690377c","subject":"Bump py-pure-client","status":"MERGED","created":"2025-05-24 19:23:16.000000000","updated":"2025-07-15 15:25:52.000000000","submitted":"2025-07-15 15:24:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950823","meta_rev_id":"b4798c34748aebe25fc31701a1328858f983b7b9","_number":950823,"virtual_id_number":950823,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-15 15:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-15 13:48:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-15 13:48:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-24 20:40:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-25 09:37:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"}],"messages":[{"id":"803939a0269dc3d1aaae794588025560e7bb20f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-24 19:23:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da8dcb09e2570e13b955786641b9d8215d18e979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 20:40:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3921c92e9e68474394362c7a5430ce94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bc4effc118450b8e51a5ecc2863877 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d99cdab8534f3e9607121203de4874 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c38dc00c34e74d5abd22d3967767594c : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1790df8733b845bd869dad0847b06fc0 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fe9831982424ca0ae832438dbef87b9 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe050d4770f14987a734f7a0524583ae : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a17b14cd924493c84693baab71831d1 : SUCCESS in 3m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b6188aab7a314a7b8e48e85e2159b067 : FAILURE in 16m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d3979bb7c4b84fed899f4dfe5d9dfb90 : FAILURE in 10m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/47956265b1964a50b5f63bc5105cae64 : FAILURE in 11m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0c11494744da4f81bf0b398fd8f724d1 : FAILURE in 19m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ceb4dda066644b42ac08dafe03d95d46 : FAILURE in 15m 17s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/26d84986f4cc47e3a9be87a307b573fd : SUCCESS in 3m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6522bd87395f4f00b11bd4b60ad7a656 : SUCCESS in 5m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3ffeef02f53b44c39889b29818285207 : SUCCESS in 3m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3e9d5caf93cc42f8b38458471a965919 : SUCCESS in 7m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/63a7fa6d296c45a4a0d044682dd0a543 : SUCCESS in 6m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/edf9305796db4df5a845db9b794f99cd : SUCCESS in 4m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fc7a6e279ba745f0b19b3189cf82ef16 : SUCCESS in 6m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/997e950510cd4533b2e83e15fc0a71ed : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/832615bc31194d6394a943361e03ea32 : SUCCESS in 6m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c9f0ad53084b495680ada195ded8bfd6 : SUCCESS in 5m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6ea808dea96b45e4a6093244198ca913 : SUCCESS in 9m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/67a07aba03e7481bbb80686f97c3796a : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a7293349ef3d4522a6e9e3cca526c80f : SUCCESS in 7m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e12ffe3d1a6c4c73b92f6c6895143dec : SUCCESS in 4m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1daa3f5afe5a4944bdfd688e0e2d54dd : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/71443d47a0a74c238fcaf329bf2bfee9 : SUCCESS in 11m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/15693bebd7e842f4a9ad4446d60ef493 : SUCCESS in 15m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/36d4e097cbf8410c89d29458e71e6b0c : SUCCESS in 6m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9c58b965aee142b7ae5217b974f8354a : SUCCESS in 11m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7712abcbfef9469d90e652366552f3bc : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3000d3912174203ac47e31e0fce74d4 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5f16f13aea64f93be87f2d45ea1a3cd : SUCCESS in 26m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3c10f8be1e0a4110910522aa874b0b2e : SUCCESS in 5m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/61616593e25545a6bd03a41f7b5b7f10 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4cf36b67e61b4aba9a46bc397564951d : SUCCESS in 4m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/56f41316f0814e46a5694af2c594343e : SUCCESS in 17m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1f9a51766b3a4320a35759be114fa780 : SUCCESS in 4m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b0c4620c2d4484e9770dd49657706ed : SUCCESS in 1h 06m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5252235eb5e445888f72f218606ea293 : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"55e84715b6309eb266c00ad43bb71e5bc9305131","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-25 09:37:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8d680068042d257b4dc966fd1fbe2fc4bcd7b3e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-25 11:36:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c3a8f224fc19a21d40e979a2628cc276b3c1839","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-07-14 19:26:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29f564bb5144339f7c5e36b646c8d0d5ea5398c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-15 12:01:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fbfd7477a5095b155b78c0512a3d645435ca28e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 13:46:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c413b6b5510b41fb859c79b91c239b44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d088294c5d34f2688e2c107b5fbe426 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5b95fc5a7f439789c6cad8daad8d48 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b033d9aa64a4504a19404741e0b593c : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac0e6f3c8cff4b5fb81800edee8cb2cf : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3edcadbc85f84af5b72373e59a368762 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da926db9f5134474a610a5775a6640e9 : SUCCESS in 2m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3f11e837c1c45cb95c9c7ea20974843 : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2d052f7bd68542ca9da17a83cca9e678 : SUCCESS in 16m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0defeb002c264879abbfac1b4034c083 : SUCCESS in 9m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fa5628e2549b49debd0e3dd3e9aa93d2 : SUCCESS in 17m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/287e792ed5ea46a38edd13aeb2cab03d : SUCCESS in 19m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/392a0b4c0ce6488f9b737f4fef906797 : SUCCESS in 15m 44s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d700b3a968f24eb1ba419c26988f96b7 : SUCCESS in 6m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4269c070232e48c7abc09948c0d232a8 : SUCCESS in 5m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2635097169b741c4aa8e30818057823a : SUCCESS in 3m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8a03abaece1a47debb3e94b7e3581276 : SUCCESS in 10m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6670c51571f74d12bf988c98ad5d7f14 : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/28fdfbe67f7341bead0560a42548f86b : SUCCESS in 6m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/707e3de383c548519c8c8da9306420e3 : SUCCESS in 6m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a4196f65b8f145bda612dc326fec87c5 : SUCCESS in 5m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72ea4434cc794c428e3bd3c4cc04c77a : SUCCESS in 5m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6169850a2c05436f9b78c76d07ec57cc : SUCCESS in 7m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/64dddaab57c54fd5a349197bd245a778 : SUCCESS in 12m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5e0e78c69b6d4f04ac96aecf3bd12ff5 : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d0555a177afb4573b9618e9defc5ca16 : SUCCESS in 10m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c07999b05c364c05861bca476ed485f8 : SUCCESS in 3m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8b6912aecc794e1b93203f43c84ef09e : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3caa7d98a8e442468b38ce3912b6e22e : SUCCESS in 9m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/792480c2a0d945dd86283f51e054a746 : SUCCESS in 22m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c369eeabdfb54a29bb26a69a9c6a19ea : SUCCESS in 7m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fdd9022a7384401aa3a62fdf94ecf772 : SUCCESS in 7m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9fac029676494acc9217161c79450266 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a14e57ac87ce4931aa7d3aa4ea4b7688 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c052eacd127741f08220421a8c4e9e2c : SUCCESS in 23m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/202dd41abc10461a93655890d474b391 : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c831ccb14788429fb34273987e31b979 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35a06187c2e44eaa8145ab2466007eca : SUCCESS in 5m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fac6da3f2d5e4844ae280ff1b88b7580 : SUCCESS in 11m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f8f4544f1c9e4244a3cbd44504dab5f6 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbe7b6e6b6e04f1893f969a5460f070e : SUCCESS in 1h 01m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4705a870705d40e29502721dc1fc348e : SUCCESS in 1h 38m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2996f5199210f927dccf46ee5619e84b0fd226d2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-15 13:48:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"644628d0d54283382f3f4a242994d146a1d7c507","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 13:48:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d5ee2eada14a0780768ee2551bf2159eaf16959c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 15:24:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e34dded7f84028aff5d41f64c9cef6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2988cab50547dba8a45f83902b333a : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e8b79c6fdb4a41b76bf2d0ecdedb26 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7aa6a7579a064228a1794813256cd459 : SUCCESS in 2m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fe0d7431d0b4ee2a24c4d42281ad8d7 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5281752702c3422a87bff450442214df : SUCCESS in 2m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1cf980ef343a4b5786e748c7769451a6 : SUCCESS in 4m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2aa4efafeb634be586eba6a65b9da83c : SUCCESS in 20m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a844209222eb4c85b216228b5c8433da : SUCCESS in 12m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/abef62f06c2647afa70aa4e6494a6ef8 : SUCCESS in 8m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c5733424292e4d379dfc37aaafec58cb : SUCCESS in 7m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e5f2d47a11524181b46233a06995d026 : SUCCESS in 7m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/46a4e0dd61d9492d8725ad1d2ae61995 : SUCCESS in 6m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1e517f7b4fe542df9bbba73b91b0a763 : SUCCESS in 2m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/53e70aa50a78485f844f0b5a1d8be586 : SUCCESS in 7m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fe8781cdf4a24f99afa15c47e355e09a : SUCCESS in 7m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a911ed6772c4ce3b31a5cee65c4c6ca : SUCCESS in 9m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/21019853ae9c4978bcb7fb33d32a32f9 : SUCCESS in 7m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/51565f74fdfd4b2f9af66de594d60e7d : SUCCESS in 8m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6bbd0879706f4c249e257501c6568b67 : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/68f22192cec4420b83d07aeb7f02599d : SUCCESS in 11m 59s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9116a8491ffc443bbf0cf6df25da3f6f : SUCCESS in 20m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/467c9e20fd5e40308a3bb6e2fc3bb3d4 : SUCCESS in 4m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/73de4772ba8548b0bf6f2edbe554e90e : SUCCESS in 9m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a005e58da4704df2b04380583a784d63 : SUCCESS in 3m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c47a408b7f2b4ae5a5a3fa0e39e86778 : SUCCESS in 2m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/960331ec45954bf7bb30dbb19a111d74 : SUCCESS in 23m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/468e7c0f0d714fd8ac93c9985d5cac13 : SUCCESS in 3m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ecb4c052e8ca4412812742059fea2eee : SUCCESS in 2m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf76b2ca87ba4b1fab751edf46ab33ec : SUCCESS in 5m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/16c9492ef5574a79a70f2c2ed01a44d8 : SUCCESS in 7m 42s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/474777b2cf5e4ed5bad03681f2cdda2c : SUCCESS in 6m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6d9c12f6d304ba3bc653a1836b4c7e4 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08c3bc8d60cf4bf3bbff214f2bbff0d2 : SUCCESS in 1h 34m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e451c9fd6b0fc3017a9b58a84d2b48625d101fbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 15:24:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b4798c34748aebe25fc31701a1328858f983b7b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 15:25:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d0e4980e2614909ae3b4609b3cd3b7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6102a3e729482ba57ffa73faafda3c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc","revisions":{"edb6d83ea50883db40843b310719018531e583ef":{"kind":"REWORK","_number":1,"created":"2025-05-24 19:23:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/23/950823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/950823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/950823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/950823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/950823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/950823/1"}}},"commit":{"parents":[{"commit":"9df4c3b73418755140fe442a389552bccaa6ea75","subject":"Merge \"Deprecate openstack/openstack-ansible-tests repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df4c3b73418755140fe442a389552bccaa6ea75"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 19:22:27.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 19:22:27.000000000","tz":120},"subject":"Bump py-pure-client","message":"Bump py-pure-client\n\nChange-Id: I906c817f144d4a1d59c1385b7f910c685690377c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edb6d83ea50883db40843b310719018531e583ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edb6d83ea50883db40843b310719018531e583ef"}]},"branch":"refs/heads/master"},"666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc":{"kind":"REWORK","_number":2,"created":"2025-07-15 12:01:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/23/950823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/950823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/950823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/950823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/950823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/950823/2"}}},"commit":{"parents":[{"commit":"1f2be26ab5a4d601127817b90c794e61cc40bd12","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f2be26ab5a4d601127817b90c794e61cc40bd12"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-24 19:22:27.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-15 12:01:09.000000000","tz":120},"subject":"Bump py-pure-client","message":"Bump py-pure-client\n\nChange-Id: I906c817f144d4a1d59c1385b7f910c685690377c\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
