)]}'
{"id":"openstack%2Frequirements~950844","triplet_id":"openstack%2Frequirements~master~I7bfd662862bde6ef2fa76a80c64f258b898be6b9","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-05-27 15:53:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bfd662862bde6ef2fa76a80c64f258b898be6b9","subject":"update constraint for python-openstackclient to new release 8.1.0","status":"MERGED","created":"2025-05-26 08:05:44.000000000","updated":"2025-05-27 15:54:47.000000000","submitted":"2025-05-27 15:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950844-new-release","meta_rev_id":"6509ae61f6f318fe22c2a57501979ff03edeaafb","_number":950844,"virtual_id_number":950844,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 15:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-27 12:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-27 12:45:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-26 09:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-27 12:45:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d201a5cb662702a4d9da64d2d9123ceac7106dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-26 08:05:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49a5277984f39040faf8cd88611d3a8f4c84adcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 09:37:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f199722de9d34cd6843bbb6589671fc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb14224b469a4e24a3a8944159d0ad4c : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c5964e15f1463db785c9407d52ba15 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63f871a309c24e39b39134fb38300be1 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae9fc22daf974ddc852fdf6071ab1360 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d818e78b5794fbebdcf16739b223574 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6daa4d03b9e744c39f4821819fcf2fc6 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/493bbb44b95640c2b37a51d79c043b6c : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/894f4b625e79455b961961acd8b31ed1 : FAILURE in 4m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e249ebaeda804a6eb74ca023fddc658e : SUCCESS in 8m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5a5fe94310e64a5fb09aa742a568e2af : SUCCESS in 16m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7a833f6ce4774bbf8f56bc68ece0fbfe : SUCCESS in 19m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/54aa705b77bf40ba959d58381f917300 : SUCCESS in 23m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fb16bc697dd645a08e87ceb813d50716 : SUCCESS in 4m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dfb8207671a64a4ea256f3027c47f028 : SUCCESS in 4m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/33ae8cb5311848be8eb39ebf7ca1b551 : SUCCESS in 3m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0d1df8e354d048c2aae6ad27330be537 : SUCCESS in 12m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ff54421c79fc4c509da9eba7f2aad65c : SUCCESS in 4m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ae6a73ec19614c63ab2fdba340317ec5 : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/619cb35be6fa4717907058e678380594 : SUCCESS in 7m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bc33189bf9d94c0bb262b9f17d8d6103 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8275df00fb4446c783414d456acc4ba3 : SUCCESS in 5m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b5a0a02af8da47c58e01e14e76f28190 : SUCCESS in 8m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/94d627c1457d4fdcbf3cf8333a79db96 : SUCCESS in 13m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/abd8d452ae434e42bb6fe9054758442a : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4c66848a47294859bb08c69d40adffec : SUCCESS in 9m 52s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/52946914819f479f9af7cd56d433d7b4 : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/df8796f487684abca2a1312ba301a9fc : SUCCESS in 6m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c293b5e47cd0455bb9b212aa798f1a8f : SUCCESS in 10m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/73ddcbf290bd485f8458c18dc00509f5 : SUCCESS in 18m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d279154cb12f41d7b8de77401d8b9b92 : SUCCESS in 9m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e8ed78c720304a38bd08ed76930a97e7 : SUCCESS in 11m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3e196f93f30e41e58b33930480a53962 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f4becd3e5b542019ba9cfe0dc249c85 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/192587d5d88d4445aa3a6d87781858bd : SUCCESS in 25m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a34bece0e0a2429ab0074f666b3140d9 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d9f54aa28a0d43e29861a4fd2c846758 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f69be1c6f66f4c6cb25fa9147fc1d8db : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d2ef846c953a4cfc99f436da80f1f45f : SUCCESS in 11m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/addae938b22b4283899faf767d847115 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8c94869dba14eb1842d7db76c09c92d : SUCCESS in 58m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a121856dbe5c4188ab25354448313ce8 : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"df479361e1231fd22b3cb2520164d0840367a797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-27 12:02:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d51c6fecdc583fa9f1c0f91e0a5af95a2c77d602","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-05-27 12:05:20.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"60a562f4b83ec5f67baba8490a8969354682379e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 12:45:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"acca41d41cf9a3e75ab56fdbabb77f83bd295b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 14:02:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/705f3fa13324496c8b4302da1fb8ef43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b5313caba947bdb08fd6fc3f1010e4 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4900c92d1b14334a42e82655283b331 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b9fae32b553478895569ebf89fa1c6b : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/103e581d8dab4669b2b6635eb76bee6f : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ecc31373127a4e3eb5efa8b1c56e83f5 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9256d2298bc247aaa1de39efb6b7a8e7 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d772390b416b4414ace06dd8720c89c7 : SUCCESS in 4m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dc648e54f56248cab944def6e2643a7f : SUCCESS in 16m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b390afaf5a5947c1a7d15f1a23670e01 : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f9748764a67a43aabaaa09cd2dbe7468 : SUCCESS in 16m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3b9c067aebc7494ab02b090f50cb6c8d : SUCCESS in 19m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2e82193d465c46948074ae86138497e4 : SUCCESS in 24m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d37e2ac5980c4e58a8cf670de6ef4576 : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ff4944d3938c4666b5f4e2451149aa73 : SUCCESS in 4m 51s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/96d36fbc1ce14a52af612cb6eda5ed4e : SUCCESS in 3m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/221195fd633c49878141e706cc66aa21 : SUCCESS in 11m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6ac1cb7fb10d4da89ba5b7241c4e56f2 : SUCCESS in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/354a5eaee72842daba4638d6bea2463b : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7f0f92ddac9442ee902b78ffe6c48fb2 : SUCCESS in 7m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/61e85656b00b43c0a74aca1d0ea3d1cb : SUCCESS in 14m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea1da7ef0b7044cf8176b5614b777a3d : SUCCESS in 6m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a0b94a880f7d4bc59eb88a46ccd816e8 : SUCCESS in 7m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/baebdcfcb30741e0a6e652120948a617 : SUCCESS in 12m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2b1f60c902814b109578c98fd1cc7825 : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/eb835d45d1e44838866177e18a10fd5b : SUCCESS in 10m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0262d0f1908c4f039746f03a25c21822 : SUCCESS in 5m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/58507559555e4a53b212da4af36dbbc8 : SUCCESS in 5m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4975655535094e26881c714859deaaab : SUCCESS in 9m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f67704dd981a47978063f2a59b348f10 : SUCCESS in 16m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/55fd80827fea48c28bcc06e1e2a422aa : SUCCESS in 8m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/944513d395934ad09ec6961ef8b37a84 : SUCCESS in 10m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4f43852967884c90b5d828602085a48b : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f32f32670fa04eff885fad249bb7227e : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c78d52e88ae647e2805623cc48199e4f : SUCCESS in 24m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f3b15d4dd2984268976b6bdd3ca666b5 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4fd67fdeaa67417e898d3a513c9e60ce : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e769fa4458474ff39300268b7cbbe265 : SUCCESS in 6m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1122b213edda48189f8eafc10efccd39 : SUCCESS in 14m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0db3ea8a9e5c4ae1b766c065c7476310 : SUCCESS in 6m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/829d233397d84c898c296f1d415f1a0c : SUCCESS in 1h 02m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8409ae00a1747ec8069feed11da173c : SUCCESS in 1h 36m 19s","accounts_in_message":[],"_revision_number":3},{"id":"cf2d254173f2d1978252b1fb62699382e28f35df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 14:03:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c30d278205d935e4abb0e8a3d2015926678a373d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 15:53:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b560cf6149754098b9a8a62d7ad9facb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af4a6e09eb34ce5883e537b64a53215 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c3c713467c4563952c01a3a6d1ebff : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9ce6d51c6b045baa56fbd4032994729 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eaa16958e2284fa8a168519470e4dc22 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6fd158eb60a4df0bd137b1cda7841e4 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac3fa770229842d2b9a687c880aa6a3e : SUCCESS in 4m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b6dc2e7f925540d6b18b0c27362ef409 : SUCCESS in 17m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3a59f33053e846288085e8fec4002111 : SUCCESS in 9m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/06987c925f964450892cebf32b0b9c6a : SUCCESS in 20m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/27668909c76c4027acb0e07228668712 : SUCCESS in 19m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/11815246752742fe8883689288f19251 : SUCCESS in 10m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e63ed5a5251d4922adb48c9840f57d25 : SUCCESS in 6m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8142da08e68a4af5bb52c865bf685d3f : SUCCESS in 5m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b3113ceac6b547bda054f75f83d7ba92 : SUCCESS in 7m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7694dc50359042898e7e91806c9a3cd2 : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07ee1780e6094f24877b074e7d937f93 : SUCCESS in 7m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f327e44921224c58ba7cae3176bd6afa : SUCCESS in 7m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0ccd13da8d24471eac0d50a15074a370 : SUCCESS in 14m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/08dc23cf04f9430bbce92dc55f7b7d8a : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9eb47daac2ea4d3285bb39870531e474 : SUCCESS in 10m 40s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/43323e41bc964e509d2ea4e975f1bc79 : SUCCESS in 22m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1e5342979ff94ad5821ca409285d48f2 : SUCCESS in 9m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b90e831f415740b3b63010de0a8654de : SUCCESS in 15m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0415c20bfee243b9a3498a89b707c2e9 : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/74f657f8e27c426cb79a5e8b8aa120a6 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1077585177eb4387ac394b2f7ec84cb0 : SUCCESS in 23m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a2db628ebb9a4e0fb580fa6b7c01e4ab : SUCCESS in 6m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ef898d10ad5a49d38bb3619c86edf107 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50487ca1518b4b9c8a00ad001b191b5a : SUCCESS in 7m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4fb102b4047547e69f4a6ca3efc73b53 : SUCCESS in 14m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/eabdb2e2cadf49d8923190549daace46 : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5de1e7faf945438786844d00a2eaf4c0 : SUCCESS in 56m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2b3034f9f80442dafc9bb054625648d : SUCCESS in 1h 42m 16s","accounts_in_message":[],"_revision_number":3},{"id":"c41b91049069106b0d6e327441aa8faac3549894","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 15:53:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6509ae61f6f318fe22c2a57501979ff03edeaafb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 15:54:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/936f3249508e49cba72051082b59c225\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c4a6a06c304450b60777830f2a1d98 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7","revisions":{"f4164589ca9731668c1e656998032551771b4377":{"kind":"REWORK","_number":1,"created":"2025-05-26 08:05:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/950844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/950844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/950844/1"}}},"commit":{"parents":[{"commit":"0697f59bec1402466cfc71e33529293ac1739e72","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0697f59bec1402466cfc71e33529293ac1739e72"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-26 07:55:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-26 07:55:43.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 8.1.0","message":"update constraint for python-openstackclient to new release 8.1.0\n\nmeta: version: 8.1.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I829c3dc9cb1fe1a84098c202b6d8f302395789f6\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7bfd662862bde6ef2fa76a80c64f258b898be6b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4164589ca9731668c1e656998032551771b4377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4164589ca9731668c1e656998032551771b4377"}]},"branch":"refs/heads/master"},"01122529115fa9b9ffbad4b9df855fac062a5bf0":{"kind":"REWORK","_number":2,"created":"2025-05-27 12:02:47.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/44/950844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/950844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/950844/2"}}},"commit":{"parents":[{"commit":"ec76c273008e3317426231cdad782622aec6e265","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec76c273008e3317426231cdad782622aec6e265"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-26 07:55:43.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-05-27 12:02:43.000000000","tz":120},"subject":"update constraint for python-openstackclient to new release 8.1.0","message":"update constraint for python-openstackclient to new release 8.1.0\n\nmeta: version: 8.1.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I829c3dc9cb1fe1a84098c202b6d8f302395789f6\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7bfd662862bde6ef2fa76a80c64f258b898be6b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01122529115fa9b9ffbad4b9df855fac062a5bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01122529115fa9b9ffbad4b9df855fac062a5bf0"}]},"branch":"refs/heads/master"},"631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-27 12:05:20.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/44/950844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/950844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/950844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/950844/3"}}},"commit":{"parents":[{"commit":"ec76c273008e3317426231cdad782622aec6e265","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec76c273008e3317426231cdad782622aec6e265"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-26 07:55:43.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-05-27 12:04:27.000000000","tz":120},"subject":"update constraint for python-openstackclient to new release 8.1.0","message":"update constraint for python-openstackclient to new release 8.1.0\n\nKept python-openstackclient version 8.0.0 constraint for\nPython 3.9\n\nmeta: version: 8.1.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I829c3dc9cb1fe1a84098c202b6d8f302395789f6\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7bfd662862bde6ef2fa76a80c64f258b898be6b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
