)]}'
{"id":"openstack%2Frequirements~951104","triplet_id":"openstack%2Frequirements~master~If215e321d7dc28e0f396acad8454d6347c2a0046","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-06-03 12:49:00.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-06-03 12:49:00.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-03 12:49:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If215e321d7dc28e0f396acad8454d6347c2a0046","subject":"update constraint for oslo.limit to new release 2.7.0","status":"MERGED","created":"2025-05-28 09:34:17.000000000","updated":"2025-06-03 12:50:04.000000000","submitted":"2025-06-03 12:49:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951104-new-release","meta_rev_id":"bec34323aad4fd5328a98082a578aef1c1a912f8","_number":951104,"virtual_id_number":951104,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 12:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-30 18:29:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-03 11:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 11:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-29 21:11:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-29 21:11:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-05-29 21:11:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-05-30 18:29:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c351ae9f60f1873335b90625eaef64af4c561c39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-28 09:34:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b629f90bbde3f12655e26302a5dd750897a3e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 11:03:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6588feb9da74464684792af021b79ff3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4168d25a6c77486ab5be5e66fcf07a28 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617921b9981b439880e289721d4df6d0 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf304f27e566428cbca270caa8565f8f : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ade627a78e44899a97b39a7b34beddc : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9860f2e1b9c34fb2acd5790724de2934 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9833d1500a354b4a871c342848eb808c : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/701dde205dcd47d6877ddca125863312 : SUCCESS in 2m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aa2892a0335642a0aa9be73bfd99729a : SUCCESS in 17m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cd147a473cb4432997e91a29386e0225 : SUCCESS in 9m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0f2fb6b933134a2e94929c0011fd0ad2 : SUCCESS in 23m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/11c45b8f237c48fca594c8ed8f2aa6ca : SUCCESS in 19m 56s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/46c7f5f9e2d94eaeb2ac4b46d6192f92 : SUCCESS in 31m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/108d82abeff04ec7b34047b51d8c3237 : SUCCESS in 7m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bf44394975c446c38bc0cfe601af82c0 : SUCCESS in 6m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/582fcc404ae247ada4ff06ee8b7b428d : SUCCESS in 6m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b5a7c49be353448f818a30d361c82743 : SUCCESS in 12m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/57be7cc4570442ec8324fdcb4947031b : SUCCESS in 6m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6a73c8b86e8f4a50840acbf1d717492d : FAILURE in 6m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/15ad5c21b12247bb93c85849d77ef241 : SUCCESS in 8m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/18699f97bf9441bca2a38da4a32322e6 : SUCCESS in 14m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d10924f12e434b32ab19d32b059548aa : SUCCESS in 6m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2f7d3cd317cd410ab3603a94e6c9b3ea : SUCCESS in 9m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d51ad699f1a14135af4aa43498385625 : SUCCESS in 13m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4f41eb2921aa448e88a5b2c5edd7b7f8 : SUCCESS in 3m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0db0ec59ccf94dd1bc8787575a6688e9 : SUCCESS in 12m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8702be0ca31d42e2bd1a743e7084f07e : SUCCESS in 5m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/93c4d2fd9625443682d043102799ea0d : SUCCESS in 5m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a3bba4ed32c74f3d84b00ed6bd5077f1 : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e25849d05c3b4d5e9ccebc077ce8f54b : SUCCESS in 20m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b4d953f3dc8149dabf5a05414b3db76e : SUCCESS in 10m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4487e855f4dd48d0bb79e70c426250f9 : SUCCESS in 15m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7379adbb4d34422782ab2b365912231c : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c558cc54cbae4dc8956ca2ad1f5082de : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/083d5922a7fe4f73b3790cf76fb63475 : SUCCESS in 23m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/34763899d75a4c3c872c7dc045aa2d80 : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aeb6e2bdd7eb4f59b8a5c2a65a2b2a30 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/615786bf9326484199f263f961a78ea5 : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6e3bc84b58684549a046fba0438fde25 : SUCCESS in 14m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0d91c1a6341146638849fb4cd502f7e2 : SUCCESS in 5m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/540c3d6c78104efbbe39835ccc6ca920 : SUCCESS in 56m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a6afbf1b6b849b482c879adbd13bf85 : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1c7a38e05297b4aa10fcdd42d3799052c2a03bc6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-29 21:11:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43de8601457693c4e978654fee33126125a0e78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-29 23:30:52.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"95cacadd3cc2b0c024ff0d6c7eeb93469b10ea51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 00:54:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/547566474104423298bde54aed15f3f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3d3bac6a314898a9c681594313541c : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9e21532e43428195ccaadb3bb26761 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e6f50197480453d81f658a449ec981c : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9d4c99bce2547fd9e560dda65ad3c49 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b592181de4a84a0b81c7485a60c35a7c : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a55dae9953240e49ee97fe7637d67c9 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ca06182976a4d56aafd9c8a8eb1fb4d : SUCCESS in 4m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e02d56fd19b5491481263395b15016cc : SUCCESS in 13m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/04bbec800206403a9b4d5bb6bb190923 : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/99e31fb6499b4b05a6d60b4c9d8aa8cf : SUCCESS in 22m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fc15c66aca734a43828d5aae1249c7af : SUCCESS in 24m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a123d3cd05a243ccb15cc67e6fccb259 : SUCCESS in 15m 40s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/00ce0743c14d4e29ac53dd34a0ebaee6 : SUCCESS in 6m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fa6e5f9880dd49df9456a8a6ed47f336 : SUCCESS in 4m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/09ec93d6196749068af1ed71a902a1fc : SUCCESS in 3m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cebccbdf0cf54abaa886bb7f07202359 : SUCCESS in 11m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a716602b98f04ca79191c4ac2c5c52d7 : POST_FAILURE in 4m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/576f681a85f4426b9eb94e104c56846c : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d98db195128b4affbc6193a9a88303da : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ecaf38ee66814cc6aeb24786e41c2085 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9fc29203540945f3b49669427f2f809f : SUCCESS in 6m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0b3bdaaee64f43b4a832fe91058774a5 : SUCCESS in 8m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/511e3ca692a04a1aa6da92b6ff336a57 : SUCCESS in 12m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3f0cc3756758436da353c08b481c623c : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e3dbdf9c508d42a797fbaa9cff9cbc23 : SUCCESS in 8m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/38c4395f7a7a44d7ad33d15d93985cd6 : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/110769caab9943b1ae4b3fe21fabc017 : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/685d1e2a3966464d93842839a6639f53 : SUCCESS in 11m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/78e9cf4767e04e8aac1c7de55c79d888 : SUCCESS in 18m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/48ea9dd4cd344b3e9cc5af02ffd9a64b : SUCCESS in 6m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/dc649659589e4647b9a550588cdaa424 : SUCCESS in 13m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/52db336ac1454370abce7d345f58264c : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/836fb711916d40afa23b680da4098df5 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08ea80977ca14d69ad02c860d701f17b : SUCCESS in 23m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e8196dc4defe44fb820b72e8a81582f3 : SUCCESS in 5m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/656d5a828037497eb905d954ec417ec7 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e381398a97ca4baa8efa91c78b69ab16 : SUCCESS in 7m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/317b82c128b44451ad9f13420ec2f182 : SUCCESS in 14m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bdd8ab3b70cf4b6087cd805c50fe53f3 : SUCCESS in 4m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9b5991f54b8490087223e66360c812f : SUCCESS in 1h 10m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28199f5d2c2b4ff6beb7afa2c8e6bf31 : SUCCESS in 1h 22m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ff2c5aa06c2ce83fd72332f8b8bddd32ebaf5225","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 06:24:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"927bfcdbcb6aa0d1aebf52aae394f1dea1391131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 07:45:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c8f50fe8b254ce18724321530876962\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87736c01ab945eea011b9b4ce843924 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110846696bef4861a377d87eea770669 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98fa15a4a0dc409eb6bc9444cd54d2fa : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8bdb1610cff44897804f0a4bf00081e6 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/247c56003a794b4eb6dfb9ae5ebb7521 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4fa4de3d6084d3dbf0529074e8dc7a2 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd074ee4ea9a4d68947b28c035523e79 : SUCCESS in 3m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7d96e80965174366bd4e74bf4585f554 : SUCCESS in 15m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dcfa57c608f64ab1b6d104f274d64deb : SUCCESS in 9m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/450f968f960b4fa5a8656491dea82d32 : SUCCESS in 14m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e998fa848cf74fb89130a999910fc610 : SUCCESS in 19m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/51c48aa50d5d44389ecf280157b401a9 : SUCCESS in 16m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ad578be4482349bca613033ab43d2d98 : SUCCESS in 4m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/74548f84e0094efd89ca708341084eec : SUCCESS in 4m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2ac5e48a5df34249a78909001f8f1423 : SUCCESS in 3m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/606502fc07ff4612a11d7ae892fd96cd : SUCCESS in 9m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f257634bcf034e6b9093db786f9807fb : SUCCESS in 3m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7e1d96db22f546d0926ba9591029fcfe : SUCCESS in 4m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1f729a12c1594368bfeaf7cf39b3cc3b : SUCCESS in 6m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/19b7cd0c8f6f4a3485575333b85a98db : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/703d0908071a47119e211545a5954e14 : SUCCESS in 8m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/db6dede109884ba2b808219b8181080d : SUCCESS in 6m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b2bafab5bd074becb8b80d5546259628 : SUCCESS in 11m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ddfe02dde19a4f779813b23a71b7bf36 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/365bf4a2e9ca41c8b936245c0f6220ad : SUCCESS in 10m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/faadf8c26f954d9ab290dfd477730027 : SUCCESS in 4m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a2e15e62ec604fed921d89582314f663 : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/95ecb7766b554f0fa872e260ac08fe6a : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8eb1f58cca764f99b5e68dae03d405f5 : SUCCESS in 19m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/da707d707f56470fa0f6833404aa1ec0 : SUCCESS in 7m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bc5a0d690d374e89aa2629959bcf90be : SUCCESS in 12m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a7416e655c634d4ea59ffa8ea09e5be3 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fe5dcafe99614ad39ba890dac89a56d4 : SUCCESS in 2m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8db82e7d0624e4f9e04365f494adf88 : SUCCESS in 23m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9b9e7690f4d143158b26964fd88a1d7a : SUCCESS in 5m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b89d1da4d8c04e88abe28ef8e17fc851 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6ab5302e66b4128ae515a69ddca227b : SUCCESS in 4m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2fcc9e7c1dc44363871fbcb6b874a61f : SUCCESS in 17m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d7a8f70f5de04a19b02409257098f1e5 : SUCCESS in 4m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07af66f6b9e9456ba9ed951ebd81a4d2 : SUCCESS in 50m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8692fce28a1d4349ab5c2fe11878edf1 : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a522c10d71ec6148254917ce59dfa4ad2b1049a6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 18:29:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dae6f8c1fc488c7215e0fe1133aa41b5ac4fff42","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-03 11:14:58.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4b6b0dca4a8381cdc10922a3dde09406405d0fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 11:15:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"15a4fe9936908d04095b9975ba51230b0d2699ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 12:49:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fbe148180824b67a14a60f8cb01b5c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e9c946960e46fc94c0cb41f607d572 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3b5dc087fd400b8cb6cdca33ec04f4 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebb99e135098484391540683aa9e955a : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9977ed0206242ecb0464ada9acecc83 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e8bf9c9301143aaa1554e30b627bb3b : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a5e8cfb89a04cbe8e1e1c70f5cf503b : SUCCESS in 4m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e49f9c40c2c44be8a4b1e330a5fc7a70 : SUCCESS in 16m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1cbd0d6f04274d1fa268e82217fd9b4f : SUCCESS in 11m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e5eb6ba30e684a8b9760256c79874a93 : SUCCESS in 17m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4d46122040e643589cfdd95e8a864272 : SUCCESS in 19m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9c6bc43e1ff74eaeae088999011d130e : SUCCESS in 11m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2fa7468ba74d4f0f8660413b1a6b85f9 : SUCCESS in 7m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/35cc1534029a4ca8ad3096e9afe8c935 : SUCCESS in 4m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e4ef9e78022b40a29d7a6b7ed961fe61 : SUCCESS in 7m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/adc597e33754443d9fc46113bdba9ee1 : SUCCESS in 16m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/729ee2743c6d414e9bc34c76fc10b574 : SUCCESS in 5m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/26d053d869c24be58c6c289168c6796b : SUCCESS in 8m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ace956c4d6244439a7749549d17b6f48 : SUCCESS in 13m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c00e1b0847284df7b1659b29bba75d89 : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/606c3de396984f8b8bc9bc91dfc70507 : SUCCESS in 12m 21s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2928270191674bfd8e4e10ae9aa3b0cb : SUCCESS in 20m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c9c3edca321844f49d3ca6f15df8ef49 : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b5c054ff73ce4d278382d54c17b20fe1 : SUCCESS in 12m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/40bb36f79fda43acaba56d1e3fa60c34 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0242aa86b4b04a27a2294b60cfebe7fa : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f7e9b5cee514fcc9208d03c6c641c91 : SUCCESS in 21m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5d8e3992adcb4ed7bd1902123ce33a37 : SUCCESS in 5m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0dfbe489e4d24ea08865eff33ea56566 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43cee66d8ea04c628d2033873e8f353d : SUCCESS in 5m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b5fc79c654ef4cab910a840b6abd1523 : SUCCESS in 13m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a9daa3b7a531403098a0c6f693f02c80 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/644e44b383594fc584340d2bcaea92e8 : SUCCESS in 1h 03m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7c8eca4e7e44a799ea604d6b8e278f3 : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3cb1a72847822df251514b394df03c967a47103d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 12:49:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bec34323aad4fd5328a98082a578aef1c1a912f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 12:50:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb365cbe0eda4337b6bcb736ad39c26f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c23e40e2dd48a5889d50040624eb70 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3c76cdea9328426c0829ebe7865c5292e7566649","revisions":{"fc5e3017f28aafa72d149633fba2e6ba645ea0e7":{"kind":"REWORK","_number":1,"created":"2025-05-28 09:34:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/951104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/951104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/951104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/951104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/951104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/951104/1"}}},"commit":{"parents":[{"commit":"631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7","subject":"update constraint for python-openstackclient to new release 8.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-28 09:24:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-28 09:24:16.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 2.7.0","message":"update constraint for oslo.limit to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I3deaa84a403c4f1b4f2df5c1cbf4e0d51ebd169f\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If215e321d7dc28e0f396acad8454d6347c2a0046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc5e3017f28aafa72d149633fba2e6ba645ea0e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc5e3017f28aafa72d149633fba2e6ba645ea0e7"}]},"branch":"refs/heads/master"},"3c76cdea9328426c0829ebe7865c5292e7566649":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-29 23:30:52.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/04/951104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/951104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/951104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/951104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/951104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/951104/2"}}},"commit":{"parents":[{"commit":"631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7","subject":"update constraint for python-openstackclient to new release 8.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/631b6b9aff1c1a1e0ee4e2043871b3519ec7b3c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-28 09:24:16.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-05-29 23:30:52.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 2.7.0","message":"update constraint for oslo.limit to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I3deaa84a403c4f1b4f2df5c1cbf4e0d51ebd169f\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\n\nDepends-on: https://review.opendev.org/c/openstack/glance/+/951351\nChange-Id: If215e321d7dc28e0f396acad8454d6347c2a0046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c76cdea9328426c0829ebe7865c5292e7566649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c76cdea9328426c0829ebe7865c5292e7566649"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
