)]}'
{"id":"openstack%2Frequirements~951450","triplet_id":"openstack%2Frequirements~master~I102b912e39a09029495fa7d282510086992e2a65","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I102b912e39a09029495fa7d282510086992e2a65","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-05-31 08:17:18.000000000","updated":"2025-06-04 20:57:52.000000000","submitted":"2025-06-04 20:56:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951450-openstack/requirements/constraints","meta_rev_id":"cdd818cca0fc1e9b1b3445ebaac0b3adad6163a3","_number":951450,"virtual_id_number":951450,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-04 20:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-04 19:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-04 19:13:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-31 10:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-04 19:13:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5981082b0ab2c6fbc3e772e491458d945bb31890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-05-31 08:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b8424d4e363547eca72c20b5ec4d0a04b71611a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 10:17:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1846864d7944985b9c2eaedf0867997\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08d1671250547afbfd701f6d45fc425 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd203c8951a543d787b0053bb9f8d4cb : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/769ae4a4bfe643a3a1e7b07924037ecd : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f129a70e5a944705b6479e0270a4b59e : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bc906d862684828aaef790648f28448 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f566305c33a4e589a8b401d03630c5d : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55ab53ba062949a1b06c0c538922577e : SUCCESS in 2m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bed08952578946e5b5205cb85bb42829 : FAILURE in 2m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/441e2d0033ac4a1b9f9aa455bca39697 : FAILURE in 10m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5464d4c991a147b59e39d61c38fba227 : FAILURE in 20m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/80aaba8fd97543e7999fb398f9a95ea2 : FAILURE in 20m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0d0a5e706fd34c00afd08bdf10f5dfd4 : FAILURE in 25m 39s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/87c3ca4a5b184ca9aa0e2aa307651d4c : SUCCESS in 4m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b731525571654a34836b432aba258da1 : SUCCESS in 5m 35s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/980636ef53dc4cd7b797df1fd84af7cd : SUCCESS in 5m 09s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5bdfcaada64d44fabb2844aaf80cb2a2 : SUCCESS in 9m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ed0c21d3b7c841a88fd59257ae295805 : FAILURE in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c59c36df7c8b4856bfefe0962d18dc03 : FAILURE in 4m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/329127e707fc4a8fb1fb1c4c0bf60dd1 : SUCCESS in 6m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6d23550ee93040cda13d4a1455416bb9 : SUCCESS in 15m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/026549b38bf84923bcd5b7b8b6b672ab : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/059c429ca1564a5f9da11b11a704ce03 : FAILURE in 3m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1ce58db6f42a4853ae58b9995afb8638 : SUCCESS in 15m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7d9a881ba46d4d07b8a64cd2aee98b3c : SUCCESS in 2m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/41679cc72d7947df82177600367e8944 : SUCCESS in 10m 04s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bf7ab6cdef2647bc96e506c12c1f2952 : SUCCESS in 4m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3a21a8c5728541a9bdd96d4c9d1f67ac : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/80ac59c8777e4dcc887e792bdbcde70f : SUCCESS in 9m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1470a431bfa648b0a37a1e24d020825c : SUCCESS in 22m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b964fae95014423aa874eeba5097091f : SUCCESS in 7m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/58005dac202c4abaaced516e9f7aec08 : SUCCESS in 15m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/485c382f2f7e422ebd4df3de3f57971d : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2173fdf9e67140ceae4fe6e76d2bd9cc : SUCCESS in 2m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8a2373f58b841b69722e14d17d02dd3 : SUCCESS in 25m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5d1f589dafc64e9aa68abb47d7f3db83 : FAILURE in 4m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1dfbcdec9e7e4cd196166231efb07f84 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9101deb76194513a4e0c5a025766cd6 : SUCCESS in 8m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/713feb0dc57c4f29b56eab606de0681a : SUCCESS in 12m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9915c74678644daa8269524087c064a6 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52b3b7a320e3474d8ce67ed0b93512e2 : SUCCESS in 51m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8d30bc6b50643388dfaef9d3ec0efe1 : SUCCESS in 1h 40m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d30255a570b340fd8da8b442e6814a5710aa9cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-03 08:00:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"055bca61d2765f36ee8bb6ee6ad7664a5674da36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 10:50:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a35c71628f3a4c8c9f1da31a54a31a47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98a078f8f634f4d91d7d086408ad588 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650db4f6566c48e2a64bde85a50d2e9f : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ff88add239343d0a0f6b956c074e2ca : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3255869b042b496a981765aaa5a96087 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87ca7cc8053442578cdfeb6ea43911c1 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ecd64ec0c5f4e83af0c8e67f3b899f1 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0fcc295aa652485eac7187975e366fda : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/09cbfe94a4624c1991268f806a6da974 : SUCCESS in 14m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/81d964b87906469b98bb7660d380826a : SUCCESS in 9m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/55d52f87715b4afcaaaff1296eaae4f9 : SUCCESS in 19m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/044ddcdcf6884723b4323f6ea176f5e5 : SUCCESS in 17m 20s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d16fa7c76d104171b21c3cbecfbc0398 : SUCCESS in 23m 03s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2fbf2c3c325a48969d5c6e098dafb2de : SUCCESS in 5m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c977bd08c0954e4891a5fd9b0c77e616 : SUCCESS in 5m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b2b60b1d9ee84ab7b3040170fdde6282 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c948778154344cd0842a335f25053a96 : SUCCESS in 13m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e52e2abc46d548968e793cfe6b203ff9 : SUCCESS in 4m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ec2cfdb0094c4c41a006484911b86133 : SUCCESS in 6m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c12ad0040a1944a5a7a0b3ab07261d6a : SUCCESS in 10m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0a14b3c654644a259616c7289a32ddb1 : SUCCESS in 13m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e60f435ade6a49e6b70b23e2437822a4 : SUCCESS in 7m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/350dba6e2783417eb574d82935bc27c0 : SUCCESS in 8m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3138d86a3ba640ce8c8389e17813d884 : SUCCESS in 11m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e87946c9c1ad4f11bf18dd7c0f7823cb : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a5e004bd4d6c480aa92f2bba423447dd : SUCCESS in 11m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6675aef1558341079175af0935b77447 : SUCCESS in 4m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bb6f4ff17d024c7b8eaf06a1facc4aee : SUCCESS in 4m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d817b67154b14d3f86628bded2cc21d3 : SUCCESS in 11m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e946d8b1af7c4607abe7070bba632b71 : SUCCESS in 17m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/137ce6c989464568950ae270e1977861 : SUCCESS in 9m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/27eb6d8081744409a05873b7a7f12317 : SUCCESS in 12m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c124cc44e2a24696a6ac212b873bb9e2 : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30a279bc0d0f49d8b2da0166d16365d2 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ee8f81ac2e140148b998ae0fb140c5d : SUCCESS in 23m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/168ff54e69c54756b89b54e064568872 : SUCCESS in 6m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/840c7697049f484b86e984895e6d6c21 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da9625647213463d97593e77f03537b1 : SUCCESS in 5m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e8cd9563be65498c9f5e17974fedd2bc : SUCCESS in 14m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f987a0391c6b433cae352316b91611bc : SUCCESS in 7m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81bdeafd00dc40a181e0f65ec400605b : SUCCESS in 1h 03m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/776733734bd14f059b5f1dcf78755577 : SUCCESS in 1h 37m 22s","accounts_in_message":[],"_revision_number":2},{"id":"00016d5e35af7ef31687181d58d58c62b24902a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-03 15:29:10.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"023b6711a2a8f5d003c57ae64363476668e47c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 18:02:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653da782f2a94bbf844f8bc6bb96c12c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36597e2c60be41d0b03b33d127ce4fdf : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8083bc29e4fc43a099a967bbb53680f3 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cbdd1d1aa424bb68aa0d32502f828f9 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43f71695c25d492e90f23d2bffc4cde4 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c25b888ad7544dfb1ac6426260c6114 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c598add1d9b54047b6167ce6af3beb00 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f655ab2a1ce5407292522ab998489d43 : SUCCESS in 2m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/60d2eb39e31d4351a07bf621044b7831 : SUCCESS in 16m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7e7fae7087b14cedb1283b20da7ecdc5 : SUCCESS in 10m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e5a8bd3f23a4484bbe0947a393b3b0bc : SUCCESS in 17m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/86b35b1c24c84b019cbfa5cfa2f86fa0 : SUCCESS in 23m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/baca3ba9624648abba6610c47c381e59 : SUCCESS in 27m 09s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/403ad26e38a14e4aabad66351571d1d4 : SUCCESS in 5m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/833ee6ef0cd54c9ab5d792c8821bf069 : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ad41eab971b747778448119282337ed7 : SUCCESS in 4m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e2f56d7707ed44ef9c500c066c55b611 : SUCCESS in 12m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2a48e6e6e6f2449eb1398c96f204c4cf : SUCCESS in 6m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/11e8abb2a11d4bdc9bc1e8b0e03aa0b5 : SUCCESS in 6m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bf7007529dd94b84a15b0be799dfed3f : SUCCESS in 8m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ddee9602dc214b4caf1361352666de21 : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9948ca996ac342eda3246386f79f1a50 : SUCCESS in 6m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cb43fb3f556541a98b551f40ec9e774d : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/edcbb52f21f242128c410fb52a5dfde3 : SUCCESS in 12m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fdd97a378fa4453da241bcb3319b09da : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/45c2aef4a21e4d3ea0ae5080e9ad4d08 : SUCCESS in 10m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f9f2522f2956463e9c4becd8ed2492f5 : SUCCESS in 5m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0945b0d113d142d882bedd2334b9b417 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a71004e6d91946dc8d291d8cfd504fb0 : SUCCESS in 12m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c2128e580fcc46bd8d9b67d95135af01 : SUCCESS in 17m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/58b020d83c44488987bfec2fb4c0e105 : SUCCESS in 6m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8a9402c652f74e678c21dc13ff688145 : SUCCESS in 13m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c0e93d454562463da839a5c29a7ddc61 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c7a86b1e4e0f47fab935a26458899a78 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b44db6b4a91347f68384f111d6b9ee64 : SUCCESS in 24m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/95dd5f22654e4d74a57cd666c4137ed9 : SUCCESS in 7m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a5914906405e43fa986f83c7fd234e53 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eec90c8c689b4405aaa12d8eef5072b3 : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/085ec060e2964de597eba88abbe7708d : SUCCESS in 15m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cb1a9d7523fd4f338401d3b76215b84c : SUCCESS in 5m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01f34f8442264e95a0e93d01f6220ba9 : SUCCESS in 51m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8892ca9a21474c87951c575630f3d32d : SUCCESS in 2h 24m 56s","accounts_in_message":[],"_revision_number":3},{"id":"357b45c956c286f04ac86b6d5c965815185c8185","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-04 19:13:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4de92840283ac1c214d85b45289de4cec7119d71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 19:13:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2a2bb85ed5457208d3ceaf4b5ae5782b73d1d48a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:56:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523011e7d26d4b959959dac410f9e554\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd90b7b1cf7b4e77b49526d93cd7fdf6 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c987d8ad6fc4fe4a2b80349f94a24c5 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/effbc434a5b749148a0f31cd382855f2 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/639cb225051f4d8d8090b3151881b6fe : SUCCESS in 3m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/028b533a9b704bbd8394675817b6ca57 : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e1ad3bf3cda94edaae1e3c5b8312e33d : SUCCESS in 2m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7707ff24a072412aa8d1a19f1eab353e : SUCCESS in 12m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54342d6bb4ac42239015473ba35689a4 : SUCCESS in 4m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/85aeff0812bd470f9e32571422593a51 : SUCCESS in 6m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d536637f46684d9d80d6d345a2135cd8 : SUCCESS in 7m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b387e84976bf45f080895a463f04f50b : SUCCESS in 11m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3760ba00c6c8469f80c74e4014b08420 : SUCCESS in 2m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fdf838c5689a470ca68d9a1219cb9cdd : SUCCESS in 6m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3ff6cd96ce2c4361ba3f79806c816072 : SUCCESS in 7m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/854b4d55747341d2ad57e817b966fa03 : SUCCESS in 5m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13ffc66709b1475690b27aabc989ebbe : SUCCESS in 7m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4a352584a50e44579e26d8c61a272d06 : SUCCESS in 8m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4b7f6b93a32d47ed9c9818a2c5fa5402 : SUCCESS in 7m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/735092ae921a468e92dd97d099b96ac1 : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c0e01c60c03c4f2092f2082b1054a389 : SUCCESS in 10m 24s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/553782f4421941d0b5e72e11504a644c : SUCCESS in 13m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ff554d5e23bf43ad9194f5b2135dc0b9 : SUCCESS in 7m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/62af49606cd14ee4a08d1c81028daacd : SUCCESS in 9m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/80cd43331bbb49f7930bb4ba473d7f6b : SUCCESS in 3m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a328c7b0bb3414eaeb0c63a56ba2225 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3bb446fc1a844878d09019bc6c21fe6 : SUCCESS in 26m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4350541b9ddf4e1fbec45d155aba2369 : SUCCESS in 3m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/18f6170317864114befc891f4c8c5989 : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/540566d2f2fc4dbcb0e22c5153332a19 : SUCCESS in 4m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b1b2c4048a44483ca87dc2c58b5c9443 : SUCCESS in 6m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bea9143b919f4526ab5c2d093f20b3ff : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00c6aea541304e3f85a431d9e3d66da7 : SUCCESS in 52m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25b8c43346e548c480d060cfde7e7aec : SUCCESS in 1h 26m 56s","accounts_in_message":[],"_revision_number":3},{"id":"411370ec4ddcc2b3b1688f411fe511140c8c0952","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:56:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cdd818cca0fc1e9b1b3445ebaac0b3adad6163a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:57:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/635a124414654d7aa6ea272ab6f9be30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94befd40ab1e472caedc72f22cc09fb5 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"602c6a5006033406f25b9f5035d8a7d206f1856d","revisions":{"9e220e3f739e613a03e5cf151d2a3938063d0a1d":{"kind":"REWORK","_number":1,"created":"2025-05-31 08:17:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/951450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/951450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/951450/1"}}},"commit":{"parents":[{"commit":"c36e570d3d645e7a51354c977783b3cd8357f91e","subject":"Merge \"update constraint for horizon to new release 25.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c36e570d3d645e7a51354c977783b3cd8357f91e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-31 08:17:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-31 08:17:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I102b912e39a09029495fa7d282510086992e2a65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e220e3f739e613a03e5cf151d2a3938063d0a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e220e3f739e613a03e5cf151d2a3938063d0a1d"}]},"branch":"refs/heads/master"},"6befc28f5d2fd4d4552506c4e655670cc1dd5289":{"kind":"REWORK","_number":2,"created":"2025-06-03 08:00:39.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/50/951450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/951450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/951450/2"}}},"commit":{"parents":[{"commit":"c36e570d3d645e7a51354c977783b3cd8357f91e","subject":"Merge \"update constraint for horizon to new release 25.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c36e570d3d645e7a51354c977783b3cd8357f91e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-31 08:17:16.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-03 07:59:18.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- oslo.limit\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I102b912e39a09029495fa7d282510086992e2a65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6befc28f5d2fd4d4552506c4e655670cc1dd5289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6befc28f5d2fd4d4552506c4e655670cc1dd5289"}]},"branch":"refs/heads/master"},"602c6a5006033406f25b9f5035d8a7d206f1856d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-03 15:29:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/50/951450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/951450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/951450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/951450/3"}}},"commit":{"parents":[{"commit":"c36e570d3d645e7a51354c977783b3cd8357f91e","subject":"Merge \"update constraint for horizon to new release 25.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c36e570d3d645e7a51354c977783b3cd8357f91e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-05-31 08:17:16.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-03 15:29:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I102b912e39a09029495fa7d282510086992e2a65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/602c6a5006033406f25b9f5035d8a7d206f1856d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/602c6a5006033406f25b9f5035d8a7d206f1856d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
