)]}'
{"id":"openstack%2Frequirements~951735","triplet_id":"openstack%2Frequirements~stable%2F2024.2~Ie7d1e04047c498d519e7670c414d1e2d51f395cc","project":"openstack/requirements","branch":"stable/2024.2","topic":"update-oslo-utils","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-06-04 06:26:19.000000000","reason":"Change was marked work in progress"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-23 14:28:53.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-06-04 06:26:19.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ie7d1e04047c498d519e7670c414d1e2d51f395cc","subject":"Revert^2 \"update constraint for oslo.utils to new release 7.3.1\"","status":"ABANDONED","created":"2025-06-04 04:27:44.000000000","updated":"2026-04-30 08:33:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"revert_of":951493,"meta_rev_id":"ba19a355b5f8eb638ca056049698387f1f270018","_number":951735,"virtual_id_number":951735,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-04 06:21:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-07-23 12:39:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-07-23 13:13:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"reviewer_updates":[{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"state":"CC"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"CC"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-04 04:27:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-23 12:39:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-07-23 13:13:28.000000000","updated_by":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"reviewer":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"state":"REVIEWER"}],"messages":[{"id":"68a7c5699889e8ea535fb5129cb18e6a13a38133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-04 04:27:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43ed25956c1e609f73a7840263aa9f07579fce9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 06:21:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96f88db92c0045cda7d9923b127c2f9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5eb78a064340e8be204a81295d33ae : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93e369451baa47f6a284a23a8ae6f9af : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b43fe2d3122c4dc6ba9c3c3b7c5c143f : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c507606da0d348b6b555f31deefde609 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a9c7da27f0b44a3a0093cde2e28942a : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2396da7be3c94c9f8777cf26e2417342 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7502514bb29b493b86fa3e531d62d51a : SUCCESS in 10m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ccdf503dacbe47b1ac4cc23c4e31a5bd : SUCCESS in 14m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/21daf33b06b8445398ddf3c35e9b2608 : SUCCESS in 10m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/208f81fed6c14d83bebb0f1ffd0b79f1 : SUCCESS in 22m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d16fc28c64c4700b8597ce4f72bc787 : SUCCESS in 24m 55s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/43c1dfb19e11430f9fc60bad1503ad39 : SUCCESS in 4m 52s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/adcb91a42ffc409590ef38dd3f6f88e7 : SUCCESS in 4m 44s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8ea599bd863c4416a70ec5470ea9aa3a : SUCCESS in 5m 33s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ce104d7de19346ebbd227df13805d790 : SUCCESS in 11m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a0026dbdf1fd4afa854734144ad2559d : SUCCESS in 4m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d4d6179723f544c9a9d2d6ec9c486ea9 : SUCCESS in 5m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3aa6a92ef0ec40aebbf068d301891742 : SUCCESS in 7m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8a58293973f0438195bc3da6b35052ab : SUCCESS in 13m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddc975921941409090b3e410252dfb24 : SUCCESS in 8m 03s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fd8eeed2b1c54671887ce6f4edc7a7fd : SUCCESS in 6m 10s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c72aa011bff64876a81f6628b7076d5d : SUCCESS in 13m 07s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4f6a6dfc0d4245f2b236c044d89e8db9 : SUCCESS in 2m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/43e7d94ac2104262b3dc7ef89479477e : SUCCESS in 6m 43s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/110140ee499a40c184ec75e03f0d714c : SUCCESS in 4m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6f1c17237a4145cda6193746e97710f4 : SUCCESS in 4m 47s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/17c8122fd5e04737a3cf52e1dfbf81b5 : SUCCESS in 11m 12s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f822ccccb4074c67af1be7736af3cc86 : SUCCESS in 17m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/791262c49f284c6cbb728838433bd9c7 : SUCCESS in 6m 46s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c3999b0834db47caba2bf87e8896bad7 : SUCCESS in 12m 14s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/836013857755477ca407fabd9360b05a : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8678a1ac5e2f4a4a8a450fb6a262662f : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88ddbadeab614aba80ed064c146dfad2 : SUCCESS in 23m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ac9de7ec25834c9a913d7a3f5be25d14 : SUCCESS in 4m 17s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c282c9eef6934dd8a77a089f3f89a665 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64dad6596951468e8aad58dd269ab09a : SUCCESS in 6m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/633718518385479c8542965e28f35387 : SUCCESS in 20m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dbdf6e298c54fa9a57c4d6288baa239 : SUCCESS in 57m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8728eb1beacb450682432c52bdbf9349 : SUCCESS in 1h 37m 29s","accounts_in_message":[],"_revision_number":1},{"id":"53d6745e202792a7c090a602453a3c5a499fcf16","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-04 06:26:19.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"fb9b9804af2ef821ab3f31a0f08f4f5c3964f728","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-23 12:39:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8f8f6877d56e85d5d1d6120a2f2b7b41b5e3d97","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-07-23 13:13:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ba19a355b5f8eb638ca056049698387f1f270018","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 08:33:26.000000000","message":"Abandoned\n\nstable/2024.2 branch of openstack/requirements transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12274f2fabe9ad1c8fddb75facc1e8766ab61cb8","revisions":{"12274f2fabe9ad1c8fddb75facc1e8766ab61cb8":{"kind":"REWORK","_number":1,"created":"2025-06-04 04:27:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/951735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/951735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/951735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/951735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/951735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/951735/1"}}},"commit":{"parents":[{"commit":"75af1eaa489b55ce43580627732d887c2ae88cef","subject":"Revert \"update constraint for oslo.utils to new release 7.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75af1eaa489b55ce43580627732d887c2ae88cef"}]}],"author":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2025-06-04 04:27:44.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-06-04 04:27:44.000000000","tz":0},"subject":"Revert^2 \"update constraint for oslo.utils to new release 7.3.1\"","message":"Revert^2 \"update constraint for oslo.utils to new release 7.3.1\"\n\nThis reverts commit 75af1eaa489b55ce43580627732d887c2ae88cef.\n\nReason for revert: Allow testing and eventually fixing the issue that led to the revert\n\nChange-Id: Ie7d1e04047c498d519e7670c414d1e2d51f395cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12274f2fabe9ad1c8fddb75facc1e8766ab61cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12274f2fabe9ad1c8fddb75facc1e8766ab61cb8"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
