)]}'
{"id":"openstack%2Frequirements~951841","triplet_id":"openstack%2Frequirements~stable%2F2024.1~Ie2fa09886206dad1ab21cbbf68e4d1b8cf2f3b90","project":"openstack/requirements","branch":"stable/2024.1","topic":"new-release","hashtags":[],"change_id":"Ie2fa09886206dad1ab21cbbf68e4d1b8cf2f3b90","subject":"update constraint for horizon to new release 24.0.1","status":"MERGED","created":"2025-06-05 10:23:02.000000000","updated":"2025-06-14 10:54:02.000000000","submitted":"2025-06-14 10:54:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951841-new-release","meta_rev_id":"47ec8508060823fa3e28750897e99d8095cf4c68","_number":951841,"virtual_id_number":951841,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-14 10:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-14 08:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-14 08:57:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-05 11:49:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-14 08:57:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"cde49e4b96d3fbe4f5bd34f4842bf8eb10a46d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-06-05 10:23:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c13b4423c69a574c132872aef34d9f424ddeb2a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 11:49:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b3c512956144333910d44ef33f97e2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d1d71274b040b9870fc2efe7b54779 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ed5fea5f3e24c5889aa289ef073ba80 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aba702c91e3d4ae69bc327ad7907e59e : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0220a1c7129240bf81edff3f0168fd3f : SUCCESS in 2m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8f86ead9af94564ba94f8e7cba7d8e4 : SUCCESS in 2m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/24ecb73cc2204f878c5a719a816c41b6 : SUCCESS in 7m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/35c90111812a4b1d8a7a55bb96045dc1 : SUCCESS in 12m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e7a5757cda24b8c8779a9d94b22a1e9 : SUCCESS in 7m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59c4cee9be0e4b3d8d8c0308d1da6ef3 : SUCCESS in 8m 47s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f27b160a1ef54a3ab42a46294fc289b4 : SUCCESS in 3m 31s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d6b0105070664bbb87055c0fca21008d : SUCCESS in 5m 22s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9f32879b4ecc40c1ad0c7421bf39c4d1 : SUCCESS in 6m 51s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/971a7d463d614aea816adf0d801cecec : SUCCESS in 7m 10s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a96bb47dcdcd4089b1ab7a5744010267 : SUCCESS in 5m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/62893464fd1949f3bfc3eb65f05a6e93 : SUCCESS in 7m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5467b39b793f478286a2772fbd9ce314 : SUCCESS in 9m 47s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/11180b8d89e4456db73796d775f6d467 : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8d045d9a31447979e9a410eacfc4dc9 : SUCCESS in 7m 10s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7b1eb3ff948d4bb69658d1eb6baded31 : SUCCESS in 9m 47s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cc0ce59536044fb49057ad14d337986b : SUCCESS in 11m 53s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/15a18041b65a43b1becb34a065ac37c5 : SUCCESS in 3m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d75008c91d4a42a08746d14a475839a4 : SUCCESS in 9m 06s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/c6a2345fd6824e09b14943b0fc9b1e16 : SUCCESS in 6m 29s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/5d9000a8031c46d982d8fdba764bae94 : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a43d9dd9408f4628a3a2a3e5b42a78ab : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5bcfab3e248d41ddab8c91fab116450f : SUCCESS in 21m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e6e70518b78b4de585fd875122f5402c : SUCCESS in 3m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6349d98f2405427d9575ef7c90f7a51f : SUCCESS in 12m 30s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/52a6eae847194b27b02b38464842729a : SUCCESS in 3m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51b1ff0ed00941be8770ce48e0365c16 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/014a72e6b26547ebba1bc0d071aae081 : SUCCESS in 22m 41s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ac83e24621f3452a8b0f2a31d8f1e12a : SUCCESS in 3m 43s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1439aa7b59f048b5a37ce06af6021097 : SUCCESS in 3m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c035d2a37064357ba21a0f32a0fab3b : SUCCESS in 5m 12s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c662c93330f74ac48edc7c1e89507f69 : SUCCESS in 7m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fe7ee99347d4ce0b1d8bfbea9833b78 : SUCCESS in 48m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f8e7151fd4a48bd9903b1ea5a921b1e : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c65517db164cc9122901d3e464376e1680d8d129","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-14 08:57:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24254d5ba4041e8e65d07a7703458d3b89435815","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 08:57:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91d36e44a699f4291248d8d294a14cab5ee8140a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 10:54:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c14114203c34803b6d17f21a236afce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea81f2b6ba4c4da19835a4a0a0601bf4 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/329056e17331454cb1b987a5a5edac5f : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6c5c323c11f4e759705b3695ffa29a4 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7eefa240a5746cd9e34313ad3f2dd26 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea25ff22aa9840b6a98074f49c700938 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dfccff226c6f41a0857c488ebd948dfa : SUCCESS in 6m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9e0d00405f5549cbbaff188875bf5096 : SUCCESS in 14m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b1d596c787f846e789648df075c115f6 : SUCCESS in 6m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/394878b7dcfb4477a3321c7c1f53aa14 : SUCCESS in 18m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6ce138e3b9094b5fbe8c3ed611252a17 : SUCCESS in 11m 04s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/93a8302f698e4888a31e62f2029045cc : SUCCESS in 5m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ef95a43a42a8428b89bbfc1ab31fd8f0 : SUCCESS in 6m 07s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/480bf283636c4937beb8bbc4f919e2d3 : SUCCESS in 7m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/37d1665f96594642b4379459dd4bc11d : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c13a264bf22a4ed6bb884f3b0437e0ae : SUCCESS in 6m 57s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d1c20ab15b2844baa087efadcdd46ab2 : SUCCESS in 7m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/4527f2b1324a4bb592fa560702223c2c : SUCCESS in 10m 28s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/94fc02aeeb0c4e989cd01cd74ae3ae47 : SUCCESS in 4m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1574a8027bc9494f9c4d285e01a0cd46 : SUCCESS in 8m 28s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/39b8b134376f44ba885a05ac075ff281 : SUCCESS in 22m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c13ee718f5704d3b8403143cf4daae85 : SUCCESS in 5m 26s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/068a45092dc04bd3a2cb43ea985f9196 : SUCCESS in 11m 43s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5fcc5703261a4103b4e292adc2b6ed56 : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c09399bf3dcf4dcb961b17f2c9a61c3e : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c7a353263084a4fa70a50df4ec997a1 : SUCCESS in 19m 17s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5d537a7224fe42d19a6aeb7069d93442 : SUCCESS in 4m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/47b09a4d68f14d5a9bdc3fbd795f37c4 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79bf7020d6e74ea58ae4b324d0e6ce4c : SUCCESS in 5m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0860eb78caa24d6b8fd7153c2cff2f13 : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c65f56eadfa94b398a42d6cb33a421d3 : SUCCESS in 44m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cd2ecfcfd0a4c898c16c52b662fb0e5 : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":1},{"id":"47ec8508060823fa3e28750897e99d8095cf4c68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 10:54:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d56805b1deb6df6a42b46fd12002945a59ad624","revisions":{"3d56805b1deb6df6a42b46fd12002945a59ad624":{"kind":"REWORK","_number":1,"created":"2025-06-05 10:23:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/951841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/951841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/951841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/951841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/951841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/951841/1"}}},"commit":{"parents":[{"commit":"b2fbe8b6cfde174cf4f35941094b5bacfb957fcf","subject":"Merge \"update constraint for keystoneauth1 to new release 5.6.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2fbe8b6cfde174cf4f35941094b5bacfb957fcf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-05 10:13:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-05 10:13:00.000000000","tz":0},"subject":"update constraint for horizon to new release 24.0.1","message":"update constraint for horizon to new release 24.0.1\n\nmeta: version: 24.0.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Guillaume Boutry \u003cguillaume.boutry@canonical.com\u003e\nmeta: release:Commit: Guillaume Boutry \u003cguillaume.boutry@canonical.com\u003e\nmeta: release:Change-Id: I916e534b26b5ef40096a6d6abd149d6c562b0438\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nChange-Id: Ie2fa09886206dad1ab21cbbf68e4d1b8cf2f3b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d56805b1deb6df6a42b46fd12002945a59ad624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d56805b1deb6df6a42b46fd12002945a59ad624"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
