)]}'
{"id":"openstack%2Frequirements~952013","triplet_id":"openstack%2Frequirements~master~I466b205cd0a5d9b572fda3b4a66057a9fc84813f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-20 14:12:30.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-06-18 07:54:05.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I466b205cd0a5d9b572fda3b4a66057a9fc84813f","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-06-07 08:41:04.000000000","updated":"2025-06-20 14:13:59.000000000","submitted":"2025-06-20 14:12:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952013-openstack/requirements/constraints","meta_rev_id":"3a264aa1d60a7257149364d9af80b794f7fc703d","_number":952013,"virtual_id_number":952013,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-20 14:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-20 12:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-20 12:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-07 10:22:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-17 09:15:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-06-18 07:54:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-06-20 12:35:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7138d066ec392f448cf9a53e4c5ffd36268894dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-06-07 08:41:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6fb3ee80b8ca4de0dd2087fc8acfeeb86314802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 10:22:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9765a7b42ae34917a8b60858062a6ae5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd36822f464d4b5a93c02667da311d2e : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5385b0e4a740b6980cc2d0ee9ecebc : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85e24bc217c7447dbe7c3144139ead59 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09f49df6ae2c41ddb2c0f88468403e43 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/341fb682ae5542dda5e208cbe36492dc : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94ca9599150d4e9dbe8cc9c111d58490 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f09bad99d4b4d95baa4dba974c7ebed : SUCCESS in 2m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/af81e2b060894fc3adc05c933babf7cb : FAILURE in 2m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9d567af62f4c42f1a064abb174a2f588 : FAILURE in 13m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2921c71f26644c7da89a59cfbbab16a5 : FAILURE in 17m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8aee1b9ab4cc46edb07804eeeeb53a02 : FAILURE in 22m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/84b7b1d2f3e24cd19c5e2837ee16a417 : FAILURE in 28m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1444938698244e0ab1a53ffee1296b55 : SUCCESS in 4m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4415bcbca2a24d85879831138c578c28 : SUCCESS in 4m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c7c4df8d273c4ee9b673f195248fbe9a : SUCCESS in 5m 42s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bd45ea0cea104f5899c14aaa6285d0c1 : SUCCESS in 11m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d28c117bb1af4c03b74a37701a1e6007 : FAILURE in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c05a6fe7bb4d490f998e433385d09700 : SUCCESS in 5m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a76b9de9f5a54883a19934cefe7e2b80 : SUCCESS in 8m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/56977856dffa49b58d47675e9a882eed : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e57d964dbc744891ba204587cbe73e64 : SUCCESS in 8m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8fb1bea4f4764cd5813e37de426514fd : FAILURE in 4m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8add8587a5ff41189fc76d3eec517ab4 : SUCCESS in 9m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9ff301157fff4e5db0d0486b89821632 : SUCCESS in 1m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/58c0ceb3d0f34a08b2170bd64261af1a : SUCCESS in 8m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/29c14e6ee04740d49dfa0e368e80daa9 : SUCCESS in 4m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6c19339bc66f4ba0a9218d997e4ec6c3 : SUCCESS in 4m 14s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/42625aeccd3a4a76b257e10dfb91f12d : SUCCESS in 10m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/75798aa4959c4177b937e2d96aa8ed5f : SUCCESS in 18m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a14e29cbef62434fbfb2428c271ceb89 : SUCCESS in 3m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9e671e272e754a32b80c36708c478a1e : SUCCESS in 11m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/869d71cdf1754ab49417338e75f9a037 : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e3806dabcb441eaba64e1ff21d744b5 : SUCCESS in 2m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f48bae180671459c9c50088902e5deb3 : SUCCESS in 25m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/22d780125bef4a68abcdf29bf998afc1 : FAILURE in 5m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1647d5ac6889490fa9a16e1eeb8a8a98 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5859be3ed2bd4d77847779ca2c628a28 : SUCCESS in 5m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/aad9ad015825434cb936e0254de08867 : SUCCESS in 15m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d53647c3eb0e4f509c26cfe59c3215dd : SUCCESS in 4m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21df8f82b1b242f2a2e617a4da93bf9d : SUCCESS in 1h 07m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26041ec2fb314060acebd7b635c39820 : SUCCESS in 1h 36m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3dfd0cddf3fc5a9af2bbdbd528c42d8248e9c33b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-06-14 08:11:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"255160c14938cbed18084c2a29770f05d448dda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 10:03:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0802ef1a6a74e7faede6b5c82020e90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7308ed28246f446a8be8d1c2ec352ef6 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2694295e8f484bc2948a15becac97cde : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15fe7c16f606480bb1fe913dcbe4a159 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bee4be0eab9144cf953bafe0e99425e0 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8c765e2e3bb4daf8ea6aa67275bd9b7 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2efc325ecbdf48c8acb07fdcaa8efc36 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d723f584e544e0989f3ab36dd9b6ecf : SUCCESS in 3m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/58fd02bdd38048dcbdb74e488bf4b972 : FAILURE in 2m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b1c5226b4361463e9d9d229d6d3eb3c9 : FAILURE in 8m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/805ee57214cd4f929dc87fdbfbf94206 : FAILURE in 16m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fa3913d825b944a9b94ddfd2a5a4a489 : FAILURE in 7m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1f3cba7526974ea3a1df031559dbc113 : FAILURE in 22m 33s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/80e4938e82254ba79c25674bdff5ecae : SUCCESS in 4m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bf6adfb9c3304767af1549b9c6d4a586 : SUCCESS in 6m 51s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d13a7ccb180f47dcabd9c276af7ec87c : SUCCESS in 4m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4e65eaa2be9e49998da0da3142ddad15 : SUCCESS in 10m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8ef82051eb8c40c3a94d0152968629e2 : FAILURE in 4m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/325e187fa7894aa3b299dd9d627e5798 : SUCCESS in 4m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f9be4a70852f4ce1a1b5d058d608657a : SUCCESS in 9m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d7627ef9a29a4c029a54589eded4fae7 : SUCCESS in 14m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4cd8465ed8d349e082782b195fd43d23 : SUCCESS in 6m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/07c2a7eadd7d410e89ffcf7fdbffea5a : FAILURE in 4m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7eb0d10c4f644a4584e18f64793c4b8f : SUCCESS in 11m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/282b32ce513e4812b5175b8f55cd1e7c : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f482cf828c8242498a447d046b7262b6 : SUCCESS in 10m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7308b895e88c493d813aa40a39559dfa : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7586d58543bb45f8a4e81b6d02e37233 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9e773ea7bd404851b57ff7b6d41cf917 : SUCCESS in 10m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/38316ffb3e3a4111a4af6278607bc8c5 : SUCCESS in 21m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98185cb7633e4c5f8729b9ffae704a3d : SUCCESS in 6m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b511e63347c74074a7f70b1ff51ab013 : SUCCESS in 14m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ca7dd078e2164380a0215de1eb7676d8 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e09091b93214433dbb838083a2212372 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c53c5c8f06147e3ba8a2e73049c400e : SUCCESS in 28m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/afaab4e8d795478c9e7cb441dde951b7 : FAILURE in 4m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/79f5a22b35d14fdf81a261a2ee22533c : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d97051c1cfd4cb2967089ddd5f752c6 : SUCCESS in 5m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5f7fe71010b74a46a17c5ea7117316cd : SUCCESS in 12m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6e07d2f2ec5a493795adb287d245bb2d : SUCCESS in 6m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50ff1b5c23524bad8123cb337f91ec11 : SUCCESS in 1h 00m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2c3900fec1040aca8cb8f9b3026ef48 : SUCCESS in 1h 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4e351796c75278ae0d5e5e05538d52c47c7ae2c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-14 13:38:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ccfc1bc3dac9f2a2c09fd865c4c276513c8eb475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 15:41:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/454aeb0308c64efa9b0564a3456b6970\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883732e2293a46b5ba96e4a87203b520 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e017c78a231748d49454da9e6e87a41b : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b742ab9d716d4fa8946de87f8b5e5ffd : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adc5f1d8f7154a7f9f1cd8cc843ee8eb : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b34f653e17a848759b03238595f86258 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb23b4d59c294513b3cbab9ca5fe2375 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/152ffb85c52f404abcb63aac4cc8917d : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aaca6bd8a66942f08c4936bf9a77a3fc : FAILURE in 3m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bc9c6a3dfb9540e2b45c4db5786f3c8a : FAILURE in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/251b08c1616643819b00eb7184347863 : FAILURE in 15m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/540cf27081774b36a5c72c3e29cd13d6 : FAILURE in 16m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/315e4fd8a6b24d37b18425c947db2f85 : FAILURE in 15m 40s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d16678fa779547e79abd4172cb6a58ec : SUCCESS in 4m 21s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/afa95538fe54499b99a711dda295b9e3 : SUCCESS in 4m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7b7bcb5d8bf842a58980dea0646c6e0c : SUCCESS in 3m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a836769aff51450589c5f5fdf7751224 : SUCCESS in 6m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4273168f9aff46f8bf43d86842ec28d8 : FAILURE in 4m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a45fff0f6f97493ca0049a3b1d4d7d1d : SUCCESS in 4m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b9996186ed5a444cb82f9ca5f8d87f43 : SUCCESS in 6m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ea6e5d2554e34e08ba7558b7bc80c56a : SUCCESS in 7m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c33ac66c19c481b8abd55dc12fe53e0 : SUCCESS in 4m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b10aa24720cc42ad88c7cc111abbc146 : FAILURE in 3m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b2acb11b25dc47beb904fa698d4a60d6 : SUCCESS in 9m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a9099d05b5d543de9f768a53978d4476 : SUCCESS in 1m 52s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5dde7872f125480bb8a3b4c650af0287 : SUCCESS in 10m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4511a6ee71244bd8958d88b8b8ed6cd3 : SUCCESS in 3m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ddfdcf9b75f448d3a0c84e1a03aaf20c : SUCCESS in 2m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ab385305ff4b4b9ea2a3aed660c70fbe : SUCCESS in 9m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e26f02dc8d7040d1b0471a1969542d00 : SUCCESS in 20m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/656c2ca30d0942c58b9be5c9e2fc9e2a : SUCCESS in 6m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/669c72bb5e044563bd35f275ca8c8647 : SUCCESS in 13m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5082f1f10a0c45828b2da1884cb10597 : SUCCESS in 3m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f049af068d314214b3efff956218d6ca : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0df328ccbee24af5ac35d635044b574b : SUCCESS in 35m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d62f46eb57574b4f8e0132c38281b0a9 : FAILURE in 4m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/deb772908e6f4315947dfd6eab011335 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c279b7752b0542f5b6227e62791de277 : SUCCESS in 5m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7fc0eb3710d14d62ab518a1d9cdfb005 : SUCCESS in 12m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8b5a5120fc6a42ea86843b263c1897fa : SUCCESS in 4m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b67d879e7c4d4d6e9adade13697d0722 : SUCCESS in 38m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14d0feedc2164a21a55a99dc2c18f950 : SUCCESS in 2h 01m 43s","accounts_in_message":[],"_revision_number":3},{"id":"54523fd728a95a853326cc43a46136e45836b3a7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-17 09:15:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"788c1a5c5b1567bae0d30be232888b7e253b2e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 10:42:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e014814873a44489bd094cf94b5a5452\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcab6a92cd14fa7a2ba6fde006e733e : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b7feb6e19e45d5b2ea614f767b0f87 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/362769e9c8ed4ef8aaf758b63fec183a : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3fa302da69f487bb7e935357c588d1a : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c6e95b9366c4cf0bc578ea68b24780d : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/491f22cefeed4d389c7b2bc6deebc6e6 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a2ad11af56b4415a6030d153c32a286 : SUCCESS in 2m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/55805575e0d24dc8adf76e39fd64b3ed : FAILURE in 3m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/369a5fb021ed4450ad1ddaaabb4d6c50 : FAILURE in 9m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2f77c5009c2d46269cff23d8252d6a33 : FAILURE in 21m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d7932f201f94ebd9eb5b494df09b459 : FAILURE in 13m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a082a5cc7cc44646b9e74a999bb56972 : FAILURE in 25m 39s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/989f178e55214a648965e20abcb978c0 : SUCCESS in 4m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6d65d0bce332439f87eb54a6ec4ea73a : SUCCESS in 4m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/affb9ba2cb8d4da5a86fbca8c0bfc46f : SUCCESS in 4m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1d485c7280cd4874a6ffec14c5eb2f84 : SUCCESS in 11m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/34ea6db40aae4f929abe238074aa188e : FAILURE in 6m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c1fe76e24ae54182b6a3a6da6c95a430 : SUCCESS in 7m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ed179dc0627548b28025550b9b377804 : SUCCESS in 7m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a5fa178a3ddc436a99162c4d3dece8e7 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c435ddece984c6e9932718416fd4c09 : SUCCESS in 5m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4ea522a784d14c22bae922bcf1e3e32a : FAILURE in 3m 36s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/00c1de2a78944ffbbb3313ad128807c2 : SUCCESS in 10m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/61523f8ef0544d228616a4ea2dbbc715 : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bdeb7329cfbc469896f7eba1aeec8dd2 : SUCCESS in 9m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c8e4c9ffc78e43839436f4425b1fc196 : SUCCESS in 4m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ea889fd49907472d997c9a3609b9df68 : SUCCESS in 4m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2f173d198c504bbfaddf7c9778ca0ad4 : SUCCESS in 9m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/67c338e68bed4174ac91a2ff62ca1e41 : SUCCESS in 22m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9dabeacbb17b425f8c7b13b5cac69905 : SUCCESS in 6m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8a6d65638a5e4008a3b52847cdd76524 : SUCCESS in 12m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/eba3c38dbe4d4fa695f6ae83b289748a : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c41b194264a8479e99207fa3b81b5251 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf051c9a77b84759a11e668c363ee66e : SUCCESS in 22m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ebd2f4046443428181ca3f47a0df166b : FAILURE in 6m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b3e308ba06164f879eade70b6c088f03 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbad9373c1904fcabff830c89fc1c1d5 : SUCCESS in 7m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2e5b51e6a68645c49a8fb34c918a24ad : SUCCESS in 17m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/57eadb08e99246349a22d59adff6154c : SUCCESS in 6m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40ae0f9cc0be4f31b667249b5f62fb8b : SUCCESS in 1h 05m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e938fa86d50e4e7aa17661156c7a0af1 : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":3},{"id":"1a74b5bf6a5ee72b2d1a2bda0c316b9fc0fd0e47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-18 07:51:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"be8a328c47172301d6cecd78157dc8da0a6890ec","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-18 07:54:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c802aa7f066a817f48196d7e27feb59bf2b40222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 09:44:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c520a3cb2b9845528a923808e8806c60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b68f73c6a714dc1827cb2b4a6ef1844 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b12f9f4bfe14989a6427828a1a7b95e : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/485fb5ff6b9e468cb320e0184dc62d5b : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6737f8d825d944159665d7008f011f44 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e451ef2503a49cab960b8b2648f9452 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85f198ff5cee46ce951316635466a739 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18d085071a844e4384cf9ab4edc9ef7f : SUCCESS in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3c9381ec1f704719b1c30b122684eeb8 : SUCCESS in 19m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8697825494ac4049b63b33d829d15b8b : SUCCESS in 8m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f573413a104e4b47b6417565ddfd5397 : SUCCESS in 19m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/201831c698b64fa7a7f06df6a41d0465 : SUCCESS in 19m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d36ca1b85d3c4edba22700fc5549f92b : SUCCESS in 23m 53s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ecba73a09d384618a9eb79b853c17fdb : SUCCESS in 5m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9942c24b6b4441efb6914ce46be7077a : SUCCESS in 5m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ed42eed8af0b46608a268963ed04ddec : SUCCESS in 5m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d8a20d30ca0b4daaa97057019a380103 : SUCCESS in 9m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cbd71e36c7244f45b0c672c657a69796 : SUCCESS in 4m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ca14d8457de245e2ae1f119aee62a766 : SUCCESS in 5m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2386f60063f5451d872102b2a85f7e36 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/17750e8fd6c648b7abf61f0b6fc55524 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95acbdad426a40e0b0ae1482c24e6f9c : SUCCESS in 6m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/af733f8a0f154d9a93682a23b55ddf95 : SUCCESS in 7m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5dca069b28a14d3287d1773bd7fc1e09 : SUCCESS in 10m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aadce186cd38430abdc7d330461e43bf : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/80eb2d98303a4b869819d60412e6ae30 : SUCCESS in 10m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2befa6466cff471787eebf9bbf0c7ea7 : SUCCESS in 3m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/529c6239d44f4165a9c65eea9cfc5919 : SUCCESS in 4m 47s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4838f827ef944737a3c7e315db1025e6 : SUCCESS in 11m 30s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/51d7969560914208b6c9a10340d71b36 : SUCCESS in 16m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3b3927e6e9f44a46a252edd1baeae2cd : SUCCESS in 7m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4fb5b3f748fb4fe39f7bcbdeee0ca254 : SUCCESS in 13m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/be4d553b3f5d4f5abe282c64d3e7ed71 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42f06f74e5f94121966d45c69cb3ffc1 : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5eb27ad776bd4ec4a2f94602b2d53ca3 : SUCCESS in 24m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/22c58a2aa0fd4158a2e8e218f32adceb : SUCCESS in 7m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5e52edaa4c5e451287ec21f4faf80dce : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c9455fd444f4c769faa58888ddd64e7 : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/09a5114171ea4e659f6c72942795aec2 : SUCCESS in 14m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/47d2b07a370c490e9123782cdca73f69 : SUCCESS in 6m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a8a8965fa5545039b6b40ab9d9cbce8 : SUCCESS in 58m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/194b61c7009c483785d4bee58f20d663 : SUCCESS in 1h 50m 06s","accounts_in_message":[],"_revision_number":4},{"id":"7b337ba719f9536b42a08784d03c625215ce7b8b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-20 12:35:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"59273e3a164de9c988b1b36f2f2cc7cf13e65e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 12:35:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3fef4a24578ccb74c02f476d5a04fd4534cc0dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 14:12:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c02b8b365a47f0a5948d11e523d2f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52fc9454f2ff443dadad25ca25d2ccb2 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae706b80eb94eba801aa6df8ba563be : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27dffa60e2ab47efb9fed1919e427a4f : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/160080d5fa614e7daaaef2067470c55d : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/394a75a9379f4557870de33df936cc20 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7cc4f8202c742549238247caad511ec : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/259df417415c4ec2a8666cad9bf876fb : SUCCESS in 11m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c2e15935a0ea4e8fa273b5c6c76676dc : SUCCESS in 8m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e165b62a3e3643e8b4723bde946e2da5 : SUCCESS in 16m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f5392996eb14455a9f735dd00eab12ae : SUCCESS in 19m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/15a53ccbd5a440d084ef940cb04006a4 : SUCCESS in 10m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e8ba390b7bde4c2fa8e7c4458ce64b3e : SUCCESS in 4m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/59dceec3826a43df82771dfef73cd5fe : SUCCESS in 7m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4d31283622ee4701b92e476644821768 : SUCCESS in 6m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ef0c840e8307486aa984800405800400 : SUCCESS in 13m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b6223f404184d62aba21d746b237e5b : SUCCESS in 7m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b3be94bc95b84f9fbd66f94ec737d61e : SUCCESS in 7m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f3c873f439af46d9b093f6a84c9e999b : SUCCESS in 12m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/41349bee964e4d878f6d3c06683d1cbd : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a75c0cac30284f5292dad0b25819ce9b : SUCCESS in 12m 11s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/64b1c36de7e4453697f04e38af961879 : SUCCESS in 16m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/233d679d62ae489fbfb7679f320d0c37 : SUCCESS in 9m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6d9bc8e679f0473898d84b33010799c8 : SUCCESS in 13m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/789552b122f741df82a6d0d992e9d6a0 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b1e3ef9bcb724c4d88b5d8cfc582204c : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b76f7bc490bd4af8ad2b6998b0fc5238 : SUCCESS in 26m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c85505d64a0f41de9f4938d39ccdfab3 : SUCCESS in 5m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/abca507270424bc6babf8142ad6f08df : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c79384a9f93e4046b9154b9df41fe6ef : SUCCESS in 5m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c356d20da19d4f9c9b882b5626fefa7f : SUCCESS in 13m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/52021ba5512247279590850ea36e3cf1 : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25ace801168a4804b893b38c0239714b : SUCCESS in 48m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f1f5da9b61044c1982373cfb93695b5 : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":4},{"id":"e5d6d2e1b0240b077332558e6fce40f25c6cc576","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 14:12:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3a264aa1d60a7257149364d9af80b794f7fc703d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 14:13:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c641fea4174d4f0f9819b68c9aa1e92a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4817a861be6145848d01a574146b30a0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"53792ea8251c126a07b80ef29cc50dc933ef660d","revisions":{"9afdae046ad321da88defca4772f0ffa0d9c35cf":{"kind":"REWORK","_number":1,"created":"2025-06-07 08:41:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/952013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/952013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/952013/1"}}},"commit":{"parents":[{"commit":"6e6cd7ed35080bbba406448d7e91e3f80df93f70","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e6cd7ed35080bbba406448d7e91e3f80df93f70"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-07 08:41:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-07 08:41:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I466b205cd0a5d9b572fda3b4a66057a9fc84813f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9afdae046ad321da88defca4772f0ffa0d9c35cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9afdae046ad321da88defca4772f0ffa0d9c35cf"}]},"branch":"refs/heads/master"},"8d96955558fab1776ef122ca1d984a1d30cdf6e2":{"kind":"REWORK","_number":2,"created":"2025-06-14 08:11:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/952013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/952013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/952013/2"}}},"commit":{"parents":[{"commit":"6e6cd7ed35080bbba406448d7e91e3f80df93f70","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e6cd7ed35080bbba406448d7e91e3f80df93f70"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-14 08:11:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-14 08:11:19.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I466b205cd0a5d9b572fda3b4a66057a9fc84813f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d96955558fab1776ef122ca1d984a1d30cdf6e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d96955558fab1776ef122ca1d984a1d30cdf6e2"}]},"branch":"refs/heads/master"},"8096933b5ce3523c15e3655e64393694b3c0f54d":{"kind":"REWORK","_number":3,"created":"2025-06-14 13:38:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/13/952013/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/952013/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/952013/3"}}},"commit":{"parents":[{"commit":"8306c31d17fe7f83bbdec752acee432d56436f8d","subject":"Merge \"update constraint for yaql to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8306c31d17fe7f83bbdec752acee432d56436f8d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-14 08:11:19.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-14 13:36:32.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I466b205cd0a5d9b572fda3b4a66057a9fc84813f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8096933b5ce3523c15e3655e64393694b3c0f54d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8096933b5ce3523c15e3655e64393694b3c0f54d"}]},"branch":"refs/heads/master"},"53792ea8251c126a07b80ef29cc50dc933ef660d":{"kind":"REWORK","_number":4,"created":"2025-06-18 07:51:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/13/952013/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/952013/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/952013/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/952013/4"}}},"commit":{"parents":[{"commit":"4627dd291b9d95463dc2dd7026892be188fef68a","subject":"update constraint for oslo.service to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4627dd291b9d95463dc2dd7026892be188fef68a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-14 08:11:19.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-06-18 07:50:21.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- openstacksdk\n- numpy\n- Sphinx\n- tomli\n- click\n- python-openstackclient\n- importlib-metadata\n\nAdd version specific pin for:\n\n- scikit-learn\n- doc8\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I466b205cd0a5d9b572fda3b4a66057a9fc84813f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53792ea8251c126a07b80ef29cc50dc933ef660d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53792ea8251c126a07b80ef29cc50dc933ef660d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
