)]}'
{"id":"openstack%2Frequirements~952541","triplet_id":"openstack%2Frequirements~master~I28c9ecb10e631c5c639586bd4160590c4aac20bb","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2025-10-14 00:58:23.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-09-11 13:20:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-13 23:16:49.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-14 00:58:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28c9ecb10e631c5c639586bd4160590c4aac20bb","subject":"Drop jaeger-client","status":"MERGED","created":"2025-06-13 02:42:13.000000000","updated":"2025-10-14 00:59:08.000000000","submitted":"2025-10-14 00:58:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952541","meta_rev_id":"997088ceab20a56fb148b4a1df45f5088017cc76","_number":952541,"virtual_id_number":952541,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 00:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-09-11 13:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-13 23:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2025-08-08 11:47:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-16 18:20:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-13 23:16:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-13 04:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 18:20:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-07-16 18:20:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-07-31 14:35:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2025-08-08 11:47:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-08-29 08:28:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"0e71ed33e0f2444ba393dba404827cf07634d78e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-13 02:42:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"944e705ce58609efa85f9d3f59b7c8d00212e72b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 04:34:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a79fcfdff9d48e8a60ae6df06ff12bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3ce36eedcf44dbb6bb4ce656f4773c : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1404a393924d93895747b700d6699a : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78cbf57edeb546668de79dc406d73c12 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e74e3cba71f4ec4a9e0c50ec80d5e41 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/830dace323674b8aaf6386b176df7e62 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77bf65068ae240dd8ca2a25d1fed0736 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/580dffad2b334a0e90798be86c063675 : SUCCESS in 3m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e599e7e544924e9f85cf1608fb57bb72 : SUCCESS in 18m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3af9e2ea36b24065a4897842cd48e332 : SUCCESS in 9m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/12226fc1c6fc4411b882985cb4c3ec40 : SUCCESS in 16m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c405ca74fc9e4ef39c567343782ae6a8 : SUCCESS in 22m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/955fc80fc1d041ce963e185b85aefc5b : SUCCESS in 9m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fcc63a19872f405cba17f0ea4715db2a : SUCCESS in 5m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/701f04f41a6440b9a2b634bf4b6378ba : SUCCESS in 5m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fc0497ed40ba43569415e8d6a5d6f54a : SUCCESS in 3m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/80ede37761ea436f982e1f0110d67076 : SUCCESS in 11m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e4d4c0ff09a84b90abb69c0ef8eb2be9 : SUCCESS in 4m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8b6d7b687bd74b6a94854a33e26d06d1 : SUCCESS in 3m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e96f59f231f84f5195fe42110313bdec : SUCCESS in 8m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3c3b09baa55b42889b300de69595a0c0 : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3743f362e0c047a8bdb2e95f3fa5ec96 : SUCCESS in 5m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/adec7180a54943a2bb9c5eff1302da09 : SUCCESS in 11m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/13b71686d18041119ac8afa9d617fe67 : SUCCESS in 10m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f591bdb51939416a88e499dfeaa19cf4 : SUCCESS in 3m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ca0e30170f644d8dbacde4b6e9da42f8 : SUCCESS in 11m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/780ddb594334444d993d8bd51d8dcc4e : SUCCESS in 4m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0ebadbb8b53c4616b207273e8b893635 : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/870e210de11d440e9a980802aa4a0826 : SUCCESS in 10m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a5cc13f978ca4bee97559f573fc0f945 : SUCCESS in 16m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2457300fed3744f6b418db99de12a5e1 : SUCCESS in 7m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8689effabb544ef696deb2930154b928 : SUCCESS in 21m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6178e105441f4283b6e38766727c0d69 : SUCCESS in 2m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f33e67c58864a4a827c003bee5c0e20 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c8b45c89ccf452c9dd689a58c3c44f5 : SUCCESS in 23m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ab85db82d8ff460fb51f7209bb7f259c : SUCCESS in 6m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/513e97147a134ed3abcc27a4daa17237 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/664146752cf24424a4deae0c9f02a57e : SUCCESS in 6m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c6678bcd21ca476f8fbc8e799a7ea5ac : SUCCESS in 10m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d598ae6927154eb38b7f251b7c2a3866 : SUCCESS in 5m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce3ffd82749046ac8d548888e98e32af : SUCCESS in 58m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ab29c3c1f9143b38d1816bbd5244eb5 : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b157fc27acac784e7cc436b4d20d4648cd1399fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 14:32:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d3cab58e2bd8178b77b6da671e9ceff9639f36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 16:11:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c1d60150f2a4d8a8743d26fb7223147\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5a8490d8ac46edb17a53eff1403426 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e004a646cdfe4badbc177313c3bd50d5 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/974d7b52ee254549a73e7b275ce308be : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/010a03402c2643feb81adf5ce5a9f760 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0686426d87104f50bc3df253692b8427 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd66a586dc594a81b50da4c917b1bcf2 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a90d98d7d884650b03f6da2079d8a29 : SUCCESS in 4m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f46de25a9593402bb6d85dbff2a9aab0 : SUCCESS in 16m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7c382d9f87c44844904e76f0fad1d873 : SUCCESS in 8m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e9c8dbb63bcd4593bc762963b263b5bf : SUCCESS in 19m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bd53201e53a543689262d5ebb4f1d9ad : SUCCESS in 23m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c3f289980953488a86c9e0506968743a : SUCCESS in 26m 33s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6debf38d1fee42e3adf5a3214a053f2f : SUCCESS in 4m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4fd37ce63455437290ba90d63bcb318a : SUCCESS in 5m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0984607d7edd4d57b5d4016065623744 : SUCCESS in 3m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f029382930454a309733943e00de60a7 : SUCCESS in 7m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e771a8ccbc7e49b99c44ee8b92cb5f80 : SUCCESS in 2m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5ab4d21ebe784c5eb94bad73cd036fc1 : SUCCESS in 6m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6662117a7fbd42f69b8baffd95ecf19d : SUCCESS in 3m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/82d264a80a4a4b3488d6cf3dc7837389 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23b3cf109fa74345bb893e5e0466c006 : SUCCESS in 6m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ed675320a08d4f64844d85f4cee8f6bf : SUCCESS in 9m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9d05a8b8c6974967a99c9b0f2e834e9b : SUCCESS in 12m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/bda89b4fec34432785aad4f4efef9c36 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9929714eb16f474598544ed147fa9da6 : SUCCESS in 8m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/72aa927767cf4cfa9a583f5e994aa903 : SUCCESS in 5m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b0c9fa97cae7473cb75bbf0e599ee9c5 : SUCCESS in 6m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b9dc0841344341d8a5a92428b9726a84 : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/90fb8dee79fe41f18fd0b7562edc0b4a : SUCCESS in 19m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1094d7604d194003be30d3178e8e7508 : SUCCESS in 8m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7305ce6dbcd64a1ea59130d9b31d9945 : SUCCESS in 9m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3124afcba25c4cf183a21698cd21469b : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d50722ac6ba4ad4838e697f9ea75bd4 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56ad452ea4764a07934c8d910308bb63 : SUCCESS in 26m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0ad965207f8e45289bc052743320b20f : SUCCESS in 7m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/655fd8bd12894d1091deb089e6fdea7b : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2797db7c4a1e471a9c42a74b303b26a5 : SUCCESS in 5m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0486c085a05c4201be3603458f1936db : SUCCESS in 13m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bd5003a4c64e4eeabb50d9cf6bb441a8 : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acefb7d75d8c4b7892dee6096f801e3a : SUCCESS in 46m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dec5b8efe9f943168ff8d6a5af530576 : SUCCESS in 1h 34m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0dfc162489026090d5c46019b801d9d14d409b80","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 18:20:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d2d2718a750f9e2079e8f2fb068cc373146bf53c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-08 11:47:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f338b740681470886f8981725cbcda797f785786","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-11 13:20:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2776a7136aebda49f29a1dd09636a3374917936c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-13 23:16:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5347377c98b5046107fe9f3dde5e857cf37494cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 23:17:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eca516cc60f72a981e36023908d45c2cff5ce677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:58:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e394029bc4f245a2b51efac33e233cd5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ffe17b2129248a49fb1aafa55ba0606 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af04d32961644d8a8865dc65a604eda : SUCCESS in 2m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a5f036dafd044f0abdf7719a08524b2 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/282468cf76dc4e46b0c270eefd0388bf : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3595285d509450c80025957c30ac7ac : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba0c866998ed4665bb452e47d36c8221 : SUCCESS in 3m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2342cf3b248b489289aadf73c0b93119 : SUCCESS in 4m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3c5371a429964f6eada7e96bcfe0b9a1 : SUCCESS in 16m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f226f000c7974a0f84a7800b41bfd31c : SUCCESS in 8m 27s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/da0361ab37b44879984a775f497f3ad7 : SUCCESS in 11m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dcd730c24010493891a34523589a7a50 : SUCCESS in 6m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bcbdd81f9a8e472791ad6fc9018dd20e : SUCCESS in 3m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/97f952f42c2d43adadab1d874c4f08f3 : SUCCESS in 8m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5168255318de4ed5bbf06e06bee9010c : SUCCESS in 18m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d130236b771d45659c500b77f4a6c468 : SUCCESS in 7m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/771922c4e09a41fc9b9df78ce0909797 : SUCCESS in 5m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9e1509570a01459fb56e97f0aa454605 : SUCCESS in 7m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a667fbe4f7c94ee7a5c0e8ed36f877e0 : SUCCESS in 1m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/164c4668e901462d956c68f671116ff2 : SUCCESS in 6m 38s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/86edfb2a12b74eb98b9d11686967f908 : SUCCESS in 16m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2000098b60c74b90aad3fb6dd763ae1b : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/97a0041b9ad14f5892be13f0ec28540d : SUCCESS in 10m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1df208f2c9f54242b8f47e99b4a47e41 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/473046784e1c481bb108fd6a334edddd : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f6169f2bcb74aa5974a3f40994d7eec : SUCCESS in 23m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8df1ce5efbd94fecbc9e533e2164a5d3 : SUCCESS in 6m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3969fd46a9804883a846dd3eb351c1bf : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12254d4b2d6549fea02db632ffed451e : SUCCESS in 7m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d3d1611fe0e24d28a712282051dfde15 : SUCCESS in 20m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2ce8d86e32fd40bf8be8dd97cf6710ea : SUCCESS in 5m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/130078e95bd84464b423e0bea6313687 : SUCCESS in 1h 06m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b89e36c7fff4520acf473cec781cc48 : SUCCESS in 1h 36m 21s","accounts_in_message":[],"_revision_number":2},{"id":"1485941247941b37acdef474a5d2fa2285d71bed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:58:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"997088ceab20a56fb148b4a1df45f5088017cc76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:59:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e05a3115c5784af7ba5a7ce663a73912\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd00e882b6a64a28a6e79332573c0535 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"18732eb7ff84910047202efec230d8b68f1fc641","revisions":{"074c2d84c53809d1954e04e457cb035c4d1d267f":{"kind":"REWORK","_number":1,"created":"2025-06-13 02:42:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/952541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/952541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/952541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/952541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/952541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/952541/1"}}},"commit":{"parents":[{"commit":"6e6cd7ed35080bbba406448d7e91e3f80df93f70","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e6cd7ed35080bbba406448d7e91e3f80df93f70"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-13 02:36:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-13 02:42:03.000000000","tz":540},"subject":"Drop jaeger-client","message":"Drop jaeger-client\n\nThe library was added for osprofiler[1]. However its development was\nabandoned in 2022[1], and now we are retiring the driver.\n\n[1] 37521f4a50e31562e7db205960e450ed0b48f423\n[2] https://github.com/jaegertracing/jaeger-client-python\n\nDepends-on: https://review.opendev.org/952373\nChange-Id: I28c9ecb10e631c5c639586bd4160590c4aac20bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/074c2d84c53809d1954e04e457cb035c4d1d267f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/074c2d84c53809d1954e04e457cb035c4d1d267f"}]},"branch":"refs/heads/master"},"18732eb7ff84910047202efec230d8b68f1fc641":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-07-14 14:32:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/952541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/952541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/952541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/952541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/952541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/952541/2"}}},"commit":{"parents":[{"commit":"1f2be26ab5a4d601127817b90c794e61cc40bd12","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f2be26ab5a4d601127817b90c794e61cc40bd12"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-13 02:36:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 14:31:44.000000000","tz":540},"subject":"Drop jaeger-client","message":"Drop jaeger-client\n\nThe library was added for osprofiler[1]. However its development was\nabandoned in 2022[1], and we retired the driver [3].\n\n[1] 37521f4a50e31562e7db205960e450ed0b48f423\n[2] https://github.com/jaegertracing/jaeger-client-python\n[3] https://review.opendev.org/952373\n\nChange-Id: I28c9ecb10e631c5c639586bd4160590c4aac20bb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18732eb7ff84910047202efec230d8b68f1fc641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18732eb7ff84910047202efec230d8b68f1fc641"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
