)]}'
{"id":"openstack%2Frequirements~953040","triplet_id":"openstack%2Frequirements~stable%2F2024.1~I9ec1db7e19a5418d2f091c5dd66810a98b4ef213","project":"openstack/requirements","branch":"stable/2024.1","topic":"new-release","hashtags":[],"change_id":"I9ec1db7e19a5418d2f091c5dd66810a98b4ef213","subject":"update constraint for sushy to new release 5.0.1","status":"MERGED","created":"2025-06-21 13:07:34.000000000","updated":"2025-06-22 21:55:30.000000000","submitted":"2025-06-22 21:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953040-new-release","meta_rev_id":"b49017570336d9dbc6736aecfc12b2b8d37ad04e","_number":953040,"virtual_id_number":953040,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-22 21:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-22 19:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-22 19:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-21 14:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-22 19:49:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d005939b8debe7208cde2b6e696c1a436cd3f97b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-06-21 13:07:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26c809e4ff4a8b5a78ce1317d719a1a7769da862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 14:11:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77481e5e11c4e23a684adb71cb29a69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f81957299a1411c9e40828b26edc9b3 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27dc9cbbe2ca49ffa3db8890d53eb1b9 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55ce7b72a31a414a9f47780197c16d53 : SUCCESS in 2m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f036d942a504802bddbdf216ca7479b : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2facea6972f7481ea9e18bfd3a8b7bd1 : SUCCESS in 2m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a5468bcb890d4d67bac64e2216c042dc : SUCCESS in 19m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/075c0f01f2494b8e9c7e907030e18665 : SUCCESS in 10m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fb9ea50c5ea9478dbf7d2c48b638fb9f : SUCCESS in 17m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bb5bab1277a14e2881273a9cc5d54330 : SUCCESS in 17m 10s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f1c45d7055eb401abf6f59ed30da8034 : SUCCESS in 3m 53s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1c5c64d6c8574e308952ca95d7c04977 : SUCCESS in 5m 59s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/27eadb0fddb64b9e89cc18a9b7e4c19d : SUCCESS in 3m 24s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/09925d139d67409eb586d5e9727be1ee : SUCCESS in 10m 36s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5186b60f5f6b4ea6a67e8119afc91cae : SUCCESS in 4m 46s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ae622fcc7a564c75a8e5b745f8d3b9cf : SUCCESS in 5m 55s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/10eb8b9901ce4d7c88e8aff0dc5c3023 : SUCCESS in 6m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/de6a97a048354f7ab11b21a56aa12b93 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c8df1795cdf46639c6afe57e59bfe0e : SUCCESS in 7m 08s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/2ae4f88cf6514c69a532d642244386a8 : SUCCESS in 7m 19s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/44ea8fc757c341668242b8e371528f4a : SUCCESS in 8m 44s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3cbf2224c60e4854be4bbd4eb387cd28 : SUCCESS in 3m 55s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b7e25ead3255489a83dd4a556e042e5b : SUCCESS in 7m 59s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/9971bd767383462eb2a6ba04213dbba1 : SUCCESS in 4m 19s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/49de6af8273e4eaaa9428905f2004e9a : SUCCESS in 4m 02s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e29193a44f2f4afdb3556788a294a366 : SUCCESS in 10m 31s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/75631a2b7798426480d4da1a9e88b4b6 : SUCCESS in 18m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/96669b1f58c44fc9841605e4ff2d329a : SUCCESS in 6m 27s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3994ae160c1b402291a632a776e36b7f : SUCCESS in 8m 54s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7163523b3f054da9b3bdbe277cd9e3a1 : SUCCESS in 7m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fee1332accc3462a89d34bab9dea0645 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7dbf70e4a9d642adb0dc7257dda284e2 : SUCCESS in 22m 11s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e346dae800ad475da6808254fb2637d7 : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0bcccfe6ce084b7eb75d9aa51e7ff694 : SUCCESS in 3m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/caceb2d57c614cc4aa44040d83265c36 : SUCCESS in 6m 32s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2d23c324e1fc4465906d64902de2bc32 : SUCCESS in 9m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/083bf9fab30e4013b42e0c93238c7c05 : SUCCESS in 55m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6af5d57a7d3c467fa441f544b75f7bb9 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a8b211ad4fecdda2b1f109b433fdfabf5e8aaf40","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-22 19:49:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"718680b382fb08a3d5636047dad76d1e8d44608b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 19:49:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6de623e17635f4f240419c9f35e74bd718add69a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 21:55:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94c13eb05e924413b23bb09568e108fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a65e953481546c68b8b6d24d98fb749 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/614c7970730140da9286c105aa506a71 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ca6a55c74164e5597e17223db0567ae : SUCCESS in 2m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/590d03b9ae8e46a5a2c741df6f4bc157 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75fa5fc0084f40d0979eaf4d4c22d44b : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/93564f8d18f9434dae1d7b70bc9a7ba7 : SUCCESS in 13m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/01a10f062eea4ec0a5a45093827d4f3b : SUCCESS in 16m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b7e4113dbe344460991a192cf195f49c : SUCCESS in 10m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8f027be4f97542ec932538c77a6f4150 : SUCCESS in 21m 57s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4d8fe26a56914b6c860ef181b034ca4e : SUCCESS in 11m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/752bffad5a974c53b030949d21e60370 : SUCCESS in 3m 40s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/beb67ebc2a074520b5d8d370b4a321cc : SUCCESS in 5m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/de13fa4b207c424aba445c56a8ceabd8 : SUCCESS in 7m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e8e38e6456c84d419998d5450f8aa775 : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/033d2ecc56a940f5ae0ed7a89471443e : SUCCESS in 7m 56s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/53b3a12e18f34997b2accca0c95632ed : SUCCESS in 6m 20s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c8333636878f462cbf478bc11f2c9a7f : SUCCESS in 12m 26s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4454ed8b893d486e880dff96201c1105 : SUCCESS in 4m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8818462d5fbb45709b98966db7b8ede2 : SUCCESS in 7m 24s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c646c72afaac4641b3a52b94b857e693 : SUCCESS in 17m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4cf82933eb8e480ca01a608b2e9286d1 : SUCCESS in 5m 36s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ce1315898766479ab9fea4d0461edd9b : SUCCESS in 13m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0da84db88d4a4f77954bc8e7f72c367f : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a17f88bf05ed4f7c9d225bfb305764fd : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/355323ab8e16464e85d4cbca9463f629 : SUCCESS in 21m 35s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b75978d9e59e465382873f40b7a88f93 : SUCCESS in 5m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e28ed2d1e32848449652a9c3eae01a08 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a5999d15c454cc6b9bb229c45f55dad : SUCCESS in 5m 20s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/bc058e1a3431475fb1d79df105523b56 : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b2c81b4b0e0485cb4696385f7a05eac : SUCCESS in 42m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ccfc924cd64406e94d592f9dfabd090 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b49017570336d9dbc6736aecfc12b2b8d37ad04e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 21:55:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3832313b21dec85f1a8bb5f9330f4d50c60b621","revisions":{"a3832313b21dec85f1a8bb5f9330f4d50c60b621":{"kind":"REWORK","_number":1,"created":"2025-06-21 13:07:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/953040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/953040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/953040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/953040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/953040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/953040/1"}}},"commit":{"parents":[{"commit":"3d56805b1deb6df6a42b46fd12002945a59ad624","subject":"update constraint for horizon to new release 24.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d56805b1deb6df6a42b46fd12002945a59ad624"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-21 12:57:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-21 12:57:34.000000000","tz":0},"subject":"update constraint for sushy to new release 5.0.1","message":"update constraint for sushy to new release 5.0.1\n\nmeta: version: 5.0.1\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I1c4db278c069a6355f5b5e8aa258ecee102f5939\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9ec1db7e19a5418d2f091c5dd66810a98b4ef213\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3832313b21dec85f1a8bb5f9330f4d50c60b621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3832313b21dec85f1a8bb5f9330f4d50c60b621"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
