)]}'
{"id":"openstack%2Frequirements~953695","triplet_id":"openstack%2Frequirements~master~Ia90b2536b1ce3cf03e27377a7a3891e728f30592","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-03 04:20:29.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ia90b2536b1ce3cf03e27377a7a3891e728f30592","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-06-28 08:23:17.000000000","updated":"2025-07-03 05:36:11.000000000","submitted":"2025-07-03 05:35:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953695-openstack/requirements/constraints","meta_rev_id":"27a8b7313bd5db0c76e1307fec248e9649f151e9","_number":953695,"virtual_id_number":953695,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-03 05:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-07-02 13:41:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-07-03 04:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-07-03 04:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-28 09:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 09:52:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-07-02 13:41:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"f2726a6fb65afdb3614dc573eb9573be277028f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-06-28 08:23:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318efd9b322731c542c9193c701c8b6dc695f61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-28 09:48:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b678bde0bc8486b9c0387f49e8c3e27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49de2336281d47b19842e9536789dd44 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ac32f990904518820df1716f36152c : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/704453806c8e41619a78b7620b003569 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d23535bf8fe24031992a1c3b3a9a2e48 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac7af102805c4a6d8791e160721a5c2d : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e109279ba74c4c838799786b78385162 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad01b0d55bb64874854bf7621edceec6 : SUCCESS in 2m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4ac419a3ff88408f9a784ef87d599e90 : FAILURE in 2m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2a17647b03bd45c29a613cc43a7ba634 : FAILURE in 9m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/63f56871448b4e739c49066920a2456f : FAILURE in 17m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/46506f18a5614736bc95eaf77e82a3bb : FAILURE in 7m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a15aa0e45bec40358179c5a870c7ebef : FAILURE in 9m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e32845bb0d4147b68991ce6769b95159 : SUCCESS in 6m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fc7e80f73803429093cc0e92c32957ed : SUCCESS in 3m 30s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/de1245d309634e8691f9c8061d7e9359 : SUCCESS in 3m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/575860592f04459297ad9446e81b4385 : SUCCESS in 6m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ae71c4b78f5d40f5981a33d9e0c64963 : FAILURE in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a7d7b569390849e98f69cb325a2a0eb3 : SUCCESS in 5m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b6d95aa1914d42f3a50d96f08b167d11 : SUCCESS in 7m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0bcfebecfe6a426abd1b0dde8fc1d784 : SUCCESS in 14m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e195d70b6afd4fbe8ed73b7f9eafacb1 : SUCCESS in 3m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ee245c672b2f497db0fd424f1df41b6d : FAILURE in 3m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/144b0a29300143afa33b1aa39f090c62 : SUCCESS in 12m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3d80bedbd6e641b2a3830829ce7190b5 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a1e9e9f67bca42938735fff413533920 : SUCCESS in 8m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/697a38a128404e82ab29a14a7e9ef748 : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e7080d8254f64791a6ec623f58caebef : SUCCESS in 5m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8d5ee76b651247f7b4e6041cd2cd340a : SUCCESS in 12m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/44fc8243be0044758a0f42b0628abfd5 : SUCCESS in 17m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6222d18a953c4f72b471fbace030403c : SUCCESS in 8m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e0c4ebedbb72405c8539496c5180db40 : SUCCESS in 15m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3e79d09c6c0e4c49b16fecd01b02ff08 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31b046a7e22f48109f99305d9cfc11b7 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a80a33cefa624f03a409f42bbe41d82b : SUCCESS in 23m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/08aa59b6c1a649ab90be8cd6dff353ea : FAILURE in 4m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dc02a36f152143fea1fd91bf3e3252d0 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d4f317785af499195c98624f7ccb858 : SUCCESS in 5m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/096aabefbdbb459b8029449fba56d398 : SUCCESS in 12m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ffce27a0f8ba48b4b15bdf1cefa386e0 : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60a6c8169c1d4aaa82c841ae54c081e5 : SUCCESS in 31m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1c0d41482fd4a7fb5484b6c5f6842fa : SUCCESS in 1h 23m 57s","accounts_in_message":[],"_revision_number":1},{"id":"df58005eb6d6196af347c1e957fdc6886e361d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-02 09:52:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ff179d02909fb3573e3ca110fa331965b358d71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 10:55:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ec3ced0e5d94833b77670c6f0c696e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b8c3a9ad25494894d5a10c34b53ed2 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a204e172b94a438ea1c9765b983fd4 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ded547d5cd9f41f5bf63925ad430d09b : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6758ba5523de45d1b69ace0b99cb1718 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e199d87a990439c962e4426cdd0553a : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f5b6e514373414bb84d0df898765515 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9f1f8ddaf694f5e9b70e42dc5e961d1 : SUCCESS in 2m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/afbb7f81ab444602a082ce25d7332cdc : SUCCESS in 17m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/17b66fe6fc544864a61add9e0efef9be : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bec13bafc5f04436a57367d85bc48c76 : SUCCESS in 15m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f35f466f44d348499c050858e0146d47 : SUCCESS in 16m 56s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2d33993de37340b1bd27f33ffc059ed2 : SUCCESS in 22m 42s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/af3dc7a6e1fb448e9776b6e450f6a8d3 : SUCCESS in 4m 26s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/184e517d5fe44acd8e51eeabf71ce6ae : SUCCESS in 4m 46s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e02bc84efb1c4f81a484c61e0823aee4 : SUCCESS in 4m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/862d51d38cd74730a9ca6a2aa3397841 : SUCCESS in 11m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fdd363942d4741a5ba074d9a7c11115e : SUCCESS in 4m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ed1b85bcb2ca475e8a9350456fc9cfa4 : SUCCESS in 4m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b1c8c94178964e9cb57bd5c0439ca976 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3083b0719b464c689aa85432cf4b33c8 : SUCCESS in 14m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe124b75525d4ce4bab43c0226cf420a : SUCCESS in 5m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e797434a87a34f4c9f4a1dfa80875861 : SUCCESS in 9m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f0ebab5f90a64e2aa9a978e865e0e8a2 : SUCCESS in 10m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/59205737672440c0ad0775a2deda21eb : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b5bfc940d0994a0aa178727f6c11e0fc : SUCCESS in 10m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/86b9858eadfd40f8b71139d36a99cf68 : SUCCESS in 2m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/421e847fdfab44b995fcc714f32aea99 : SUCCESS in 6m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ce98d45af8724c839b8dffb816d4f7c4 : SUCCESS in 12m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a1fcbcbe48944960948e2fdcacfc4104 : SUCCESS in 18m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4617480b2a6b4c9690c0c9a52e1304d6 : SUCCESS in 9m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3062ae4518604f4cacfab9ba6b0ff102 : SUCCESS in 11m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/eea3f8cc49f94d728c11de58c1d27fd3 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7cdc3bba97e7415597fd7acef7b8ace2 : SUCCESS in 2m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c607d28bde04cacbd13e1ac0705c7f8 : SUCCESS in 22m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/588ba9bc111f4bc48f3db632d849b5f7 : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aecf13868fd749178434019ed99e7c02 : SUCCESS in 2m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7fbb08963a05468b9715c00c6e1bb829 : SUCCESS in 7m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9b3c19938fe94259a751ba495bd1f5e3 : SUCCESS in 17m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/39942acf06cb463a9b4a9b6b14d4a296 : SUCCESS in 6m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4ad60f7d1034cf48370e83f2ea4a74c : SUCCESS in 57m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1be1cd13675345c18f8245678027732a : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":2},{"id":"272ab590987a08c62c3ba4dc411fcb5f3272d473","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-02 13:41:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32318454185a16d15916146fb7641c381c6f3a80","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-03 04:20:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dfa090d47cebd3707c8ec5d5a297eab02a338932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 04:21:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bbcaf966ef2176210c43545506e81c3c908edec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 05:35:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5918656ef615459380133112d1fe5da9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ad49053f3f47818fe60828af6113f9 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee19a3c6f5848ac86e75e3278320761 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef4481507f5149eda97f672c48fbbe17 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36509131490b44f084e3b76be86c6e7c : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b981365d0b54552b4c345981330ca43 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e6e601b00c7413ea44bec5ec6c8e8f3 : SUCCESS in 3m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/908f25c4909646b1934d4335df014bd4 : SUCCESS in 6m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/35a1062d001946a2b80e2bf104e54374 : SUCCESS in 10m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0fb49b7e16794941b0c64dcbfa93c344 : SUCCESS in 17m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7f0153abe57a4c9db65f84be45d0bd46 : SUCCESS in 14m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6fed635cfc7c47568ac141d6db870218 : SUCCESS in 10m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c3a92297cd8145c9ba67736c06a3fecf : SUCCESS in 7m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/018833fc26de48448c28ac94d806ea0f : SUCCESS in 4m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/179a81355fd44cc3ae42677f63a753de : SUCCESS in 9m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/91734e78a3ae4fb69d58850f39984417 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af6bb4180cc043d4bf78a6925789deec : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6e5e4866f16c4c1fa1a59829ece8bf16 : SUCCESS in 8m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/58942a45c3b64e459124f3e3c8ede96e : SUCCESS in 12m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a28b6d4dfcdf4f089749347c0ba4f3c6 : SUCCESS in 3m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/82f01941d9d34aad9f574864c2832363 : SUCCESS in 9m 03s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/de2072790224463baab9bc5a41be94e9 : SUCCESS in 14m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8f2d9f94068242c6b58aa51f8fd84e0f : SUCCESS in 6m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9b371152433f4db98861993c5ff82658 : SUCCESS in 10m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/af979aac878647e8ad99624739743dcf : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3bde08560814a00a2f7f104d3f973a4 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/170b7862038649e4bd30f32761867d66 : SUCCESS in 25m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/08c53d36df0a47cab8ee4e9ad693be30 : SUCCESS in 6m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6f8669ed63274bc69e9a7ac60f6fc311 : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98ae39d23c984f82b75d7bf68d923efe : SUCCESS in 6m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/efc79c8ea3dc4641a93939d4639b4346 : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cb06ce382c9749de99b86e88b51f1188 : SUCCESS in 4m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/caf26a318ced4bafadba6820ef6f709d : SUCCESS in 44m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edb3c5f6b8034102a3c805b59dae8aa3 : SUCCESS in 1h 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6294513dc1966100e04e84bdb3aea6f48090c6ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 05:35:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"27a8b7313bd5db0c76e1307fec248e9649f151e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 05:36:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f84157199e4dc5b99fb07dd56e6e85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09c48d64a7e4a8480b2cc6940812f2b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6dc821ff640d06e71132059ee94a2acd7cd9ebf3","revisions":{"9dcb4ae74256188b81995198b488ceb794b9f00c":{"kind":"REWORK","_number":1,"created":"2025-06-28 08:23:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/953695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/953695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/953695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/953695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/953695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/953695/1"}}},"commit":{"parents":[{"commit":"d6733de49efe2c1cecce997516e7eaad3d29cd88","subject":"Bump eventlet to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6733de49efe2c1cecce997516e7eaad3d29cd88"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-28 08:23:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-28 08:23:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia90b2536b1ce3cf03e27377a7a3891e728f30592\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcb4ae74256188b81995198b488ceb794b9f00c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcb4ae74256188b81995198b488ceb794b9f00c"}]},"branch":"refs/heads/master"},"6dc821ff640d06e71132059ee94a2acd7cd9ebf3":{"kind":"REWORK","_number":2,"created":"2025-07-02 09:52:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/953695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/953695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/953695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/953695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/953695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/953695/2"}}},"commit":{"parents":[{"commit":"d6733de49efe2c1cecce997516e7eaad3d29cd88","subject":"Bump eventlet to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6733de49efe2c1cecce997516e7eaad3d29cd88"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-06-28 08:23:16.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-02 09:50:16.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- doc8\n- async-timeout\n- exceptiongroup\n- daiquiri\n- cliff\n- numpy\n- Sphinx\n- tomli\n- click\n- openstacksdk\n- python-openstackclient\n- importlib-metadata\n- scikit-learn\n\nAdd version specific pin for:\n\n- txaio\n\nPin versions for packages that are not ready for updating:\n\n- py-pure-client\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: Ia90b2536b1ce3cf03e27377a7a3891e728f30592\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nGenerated-by: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dc821ff640d06e71132059ee94a2acd7cd9ebf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dc821ff640d06e71132059ee94a2acd7cd9ebf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
