)]}'
{"id":"openstack%2Frequirements~954016","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I7a960aa06d37ea1e53735c17b5267b19a47f7e28","project":"openstack/requirements","branch":"stable/2025.1","topic":"mtls-epoxy","attention_set":{},"removed_from_attention_set":{"30126":{"account":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"last_update":"2025-07-07 05:30:45.000000000","reason":"\u003cGERRIT_ACCOUNT_30126\u003e replied on the change","reason_account":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}}},"hashtags":[],"change_id":"I7a960aa06d37ea1e53735c17b5267b19a47f7e28","subject":"Bump oslo.cache in upper-constraints","status":"ABANDONED","created":"2025-07-03 06:57:36.000000000","updated":"2025-07-07 05:40:18.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e4f58e219f554d7002da9201674b37e86ff4d6d3","_number":954016,"virtual_id_number":954016,"owner":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-03 07:58:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2025-07-06 08:25:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-07-07 05:30:45.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-03 07:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-06 08:25:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3cf8a5cdf53245ee76ce5d0d81d88db98fe7c8a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-07-03 06:57:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e10946d5ab2087e70fb24ec7f7182e4632237454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-07-03 07:02:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"19c33c0cbb0c27af9c268aa2507edef9219bfdb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 07:58:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0de82d1951047d9bc319204ae05cd89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb451c379b00437a942acf442293418f : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53e1c74be86f458e93bacff4b8ce3fc0 : SUCCESS in 6m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e284542a2cac46ee8f4e5ffd254d0e27 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92277feea944403181253a7eab9c4cf2 : SUCCESS in 2m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8e8651aa88143569ebf52c3bb1ab9de : SUCCESS in 2m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e456bec8eab448838a788f91b42ac8c2 : SUCCESS in 6m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/91df66933b1e4363bbcdb0ecf92581c9 : SUCCESS in 7m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/51ea03e878e34a7ab611aea324146c7f : SUCCESS in 19m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/69bff9314cd64cb9b2ceaf9c9ba24050 : SUCCESS in 13m 19s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/754d4b97a5c4464ba2e4b0f40b4fae7f : SUCCESS in 6m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7e20053e84644182aa5d92d58017c4d2 : SUCCESS in 4m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e3c84416f3d34007b0da5567e507d0e2 : SUCCESS in 3m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b83af31768604edeba3e47839a8b4176 : SUCCESS in 12m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ffad262d7dd245fc8856eac43cd9eaee : SUCCESS in 4m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/db4b429776dc4f85b59296f71daf3f36 : SUCCESS in 4m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ba4a0e4aedb64a2688b5dac99e3e7ee8 : SUCCESS in 6m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a56d60bb188f4b2581c95044be689a02 : SUCCESS in 14m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e77ec40db2f54bcfb8927aa0389b47c6 : SUCCESS in 13m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4ec4165715a54e568ff449ae1091412b : SUCCESS in 6m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8e1341650260408cb4f1730b38953116 : SUCCESS in 9m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/911b106c5f764fda813f48ac5f059c50 : SUCCESS in 3m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d6806bab1824482fae3db34147f1228d : SUCCESS in 9m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0e0683d3fcf94ba8880c32a2177dac83 : SUCCESS in 5m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/86bbeb1e7fee46f8ab5ea81f3803ac6b : SUCCESS in 6m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7b57244c081a41c3a2714144a1d3f958 : SUCCESS in 9m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4ae1e74b73734574bb932f3931d28508 : SUCCESS in 14m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5598f23e52a3473dac3729861c0e902f : SUCCESS in 7m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3967889fc8f54d8886fd8b2cd39c26fe : SUCCESS in 10m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0bda5cd7a4ba40a7b42d197c01f132ff : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7afa40e094d343dfbaedc37288c88d91 : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3611eda2180242a89377f79ba0ca0ad1 : SUCCESS in 27m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3c2b589c2fe1421c9d1766a4ce882ad9 : SUCCESS in 7m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cba84dd2f8fc423bbf9a63854c3bc24f : SUCCESS in 3m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bfacaba06614c13a8b425cfc90aba39 : SUCCESS in 4m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dd6e316d358747938495890f5d4035d4 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf92798a04c4449993bd417d5db0907b : SUCCESS in 53m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70e5194029fe44ae8322e7d51af8f3da : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":2},{"id":"583f8ef30ed43483635e88607d87aa778a215ba7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-06 08:25:12.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c5cbe22956df75f4c75b2bc6651f2f19f5409f5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-07-07 05:30:04.000000000","message":"Topic set to mtls-epoxy","accounts_in_message":[],"_revision_number":2},{"id":"9ae0a93bd7b2b4ee94008bf2035fbc431456e5a6","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-07-07 05:30:45.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4f58e219f554d7002da9201674b37e86ff4d6d3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-07-07 05:40:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88f67590517002c3ff7b96f5d29738cec425b107","revisions":{"2fe6acae694bc3f0b10493c381d2b077d1dca511":{"kind":"REWORK","_number":1,"created":"2025-07-03 06:57:36.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/16/954016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/954016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/954016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/954016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/954016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/954016/1"}}},"commit":{"parents":[{"commit":"e31b331399de82c2277635b0c2ee2a6a271cd229","subject":"update constraint for python-zaqarclient to new release 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e31b331399de82c2277635b0c2ee2a6a271cd229"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2025-07-03 06:56:56.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2025-07-03 06:57:21.000000000","tz":120},"subject":"Bump oslo.cache in upper-constraints","message":"Bump oslo.cache in upper-constraints\n\nChange-Id: I63b04b514aa4b927946f85e2fc120533a99326a4\n\nChange-Id: I7a960aa06d37ea1e53735c17b5267b19a47f7e28\nSigned-off-by: Luca Miccini \u003clmiccini@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fe6acae694bc3f0b10493c381d2b077d1dca511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fe6acae694bc3f0b10493c381d2b077d1dca511"}]},"branch":"refs/heads/stable/2025.1"},"88f67590517002c3ff7b96f5d29738cec425b107":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-07-03 07:02:34.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/16/954016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/954016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/954016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/954016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/954016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/954016/2"}}},"commit":{"parents":[{"commit":"e31b331399de82c2277635b0c2ee2a6a271cd229","subject":"update constraint for python-zaqarclient to new release 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e31b331399de82c2277635b0c2ee2a6a271cd229"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2025-07-03 06:56:56.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2025-07-03 07:02:18.000000000","tz":120},"subject":"Bump oslo.cache in upper-constraints","message":"Bump oslo.cache in upper-constraints\n\nChange-Id: I7a960aa06d37ea1e53735c17b5267b19a47f7e28\nSigned-off-by: Luca Miccini \u003clmiccini@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88f67590517002c3ff7b96f5d29738cec425b107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88f67590517002c3ff7b96f5d29738cec425b107"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
