)]}'
{"id":"openstack%2Frequirements~954219","triplet_id":"openstack%2Frequirements~master~Ie2b580722611b2c929fe09bff1d97f12749b559d","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-07-11 00:16:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie2b580722611b2c929fe09bff1d97f12749b559d","subject":"update constraint for python-blazarclient to new release 4.4.0","status":"MERGED","created":"2025-07-07 10:42:19.000000000","updated":"2025-07-11 00:17:31.000000000","submitted":"2025-07-11 00:16:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954219-new-release","meta_rev_id":"93703e25e142212666ecdd3462c1090adb5d932c","_number":954219,"virtual_id_number":954219,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 00:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 18:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 18:45:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 12:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 18:45:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"54a3a90ad301f1157c2b4f92417526d7363815e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-07 10:42:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"396674089c5e25d60516682973d248d47fcf2d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 12:22:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87e77824dcab4ffcb2b52ed762305dee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9217d75a5347a098abc018ec37a0b3 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace78eb8751649a88ccd9e59482ba3ad : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e012c9c30a4d4c18933a7c5a66921603 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9af6f9a4fdd49f2a9abb0ce0fbbdc44 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddb70319162748dcaa48ce1470e09be5 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/063e9df374cb4f46a858b31a430b2025 : SUCCESS in 2m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0906cb61fe964845ab76cd09fa0d998e : SUCCESS in 4m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9f7659ec49044fa19ef404c6db9191df : FAILURE in 3m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a1788834ef2b4f3f937cb0324c50373c : SUCCESS in 10m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/56da53bc2cc14446971c1a3d67f050a4 : SUCCESS in 15m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d9cd7c4e214b44f99ac7c8a9d6c4aee9 : SUCCESS in 18m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e9a53677753246e3a3416db9e883e5ed : SUCCESS in 25m 19s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/64c0249427eb43698a8230d1bf5c8e27 : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dfedd7a018094c0baa04debb6f72e387 : SUCCESS in 4m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dbd8ebed22674f2b80dbe8e9e74b42ff : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cc0c1c471c3e4f6aaa9bdf723e60b434 : SUCCESS in 6m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9e41378e3c624af593b179ec99f7636d : SUCCESS in 5m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/39db274fba4b4335a5b7087d0b7b8ca9 : SUCCESS in 3m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4e580f59a0494fafa0b863e3b6d65b35 : SUCCESS in 8m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d84cc42a8dfa4b359ff1a2a9686ae662 : SUCCESS in 14m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73b9f0e62d52479d81b936e9d342f52d : SUCCESS in 4m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/29d1a5b17099404ead41a70ae6586b47 : SUCCESS in 5m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6159aa3c073a4fa5a44715d884edd21e : SUCCESS in 7m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8c2357a182a54a26962b61ec837236a5 : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/921f4e1a15924a50b48185a0c174a7a2 : SUCCESS in 9m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4b19ca58b51f4eecb334c923806c1fed : SUCCESS in 4m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/23651a39795146539d4be00602c51c48 : SUCCESS in 2m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d344ad8f799b4aa9b4b6fff9117190a7 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/39acc123fde343b99e4637d38e60fce2 : SUCCESS in 21m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6a17ef4ebf154da1b4613eb220f0f63c : SUCCESS in 4m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3372175b95f240c387aa2fd97f25c752 : SUCCESS in 12m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cf7b26dd66df413382da5a9ac314d863 : SUCCESS in 2m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4651fd9161a248b3aaa109d3d45b3757 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f559d8ce0da249d99d6835550cdeabd8 : SUCCESS in 29m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d9e41905a07142e88e1b7348de700b94 : SUCCESS in 7m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6f9fd92a739e4dffab3e4ab7413af054 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42f95f17c0314bb1b412c3d5b0800713 : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e09212f2dd924086892e30c1b3ef0406 : SUCCESS in 12m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d07e05f9fe32429ab84c99c907006c81 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/078f4da263e542be93ce0ba5fc9279ce : SUCCESS in 50m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fa45cdd225942a09c61a6f31909ccf0 : SUCCESS in 1h 35m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3d24a54034a03a941d1b73b1dc65286a175ab900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-10 12:24:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a288001657a3d537f899958148a0e77e6d13a84b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:32:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21236ac998b84b929aeffdf443909a11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd13c88d152c4abfa7e5a599267f54ff : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae256d7ac364433935755641b04a2ea : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/594818d1558d4df69c228f5c775bd985 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e4a19f3c093420d8cea1e4bce8d252b : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4de3d9595c7648f3bddafd9b8d305ee0 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d54cadef6434a39aed116b92c074f01 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2f6ea13e4e048daaa4d1cfc031ccfe3 : SUCCESS in 4m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0f5649c0217c4ea8868b37cde4c7c387 : SUCCESS in 16m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7b75c688520343938099e555a86fb7b5 : SUCCESS in 9m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/613545d80fb444458e633b011b61b8ed : SUCCESS in 17m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/32435855c72647e08a1679b731a51210 : SUCCESS in 18m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/685b75cf7faf454ba7e67e111d0f40f0 : SUCCESS in 28m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/78f22c64dcf247a296d297fbbe019f14 : SUCCESS in 6m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/00183b53d8324dac979ba676934321f7 : SUCCESS in 2m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/cc8b77b1b29648408469c1756bd8eadd : SUCCESS in 2m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5f6647780a7549f585ef6d342bb6d420 : SUCCESS in 11m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4b5dcd83259e45e89eea98a545ed67da : SUCCESS in 8m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/93822d452d454dd88ce98c5724072ce8 : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/48ea202529644ee4b74ac437c6b89ad1 : SUCCESS in 8m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9a62cb04fecb4af19171162da096af2e : SUCCESS in 15m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/339d8f0ef496451497ca03dfc00bd1bb : SUCCESS in 6m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/aa8518ccf3b24619abde5e7086032ee3 : SUCCESS in 4m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e3f912b6e11948e9b0828fe5d216b24f : SUCCESS in 9m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cc354fd611cd4fe8b9b513cee466e05c : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/391a5282a241404bb59a9d08ef54c4ba : SUCCESS in 14m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fced21f7f499482592cb2be6662424c5 : SUCCESS in 7m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/fe4aedf726864b5ca469b106d7a2a950 : SUCCESS in 4m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/48a0aaba827f43548b23651f9c572da7 : SUCCESS in 10m 35s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9a1e776e9b3948f29ac90e846988d0a2 : SUCCESS in 17m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9e00db8276834ce3acb7f82cb176c625 : SUCCESS in 8m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4c3aa8a789274769a2cf3e943ba96bac : SUCCESS in 14m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4ae50e6d1c024b96bb3dd1b91afc21d6 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/50481feb11734510aa3a3b98397c72d8 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a3246ba0d1f4b5f97820d99bb6644a7 : SUCCESS in 28m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/98b4e24fb10f4ca6944f502a63ce144b : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8c786878954e406b9a958dfda572d4c9 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97ae73fd77d041e58ce8a88cbd9b792e : SUCCESS in 3m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f76f11031d8a4096b5a3e26c649afe0b : SUCCESS in 15m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fdb005e7bab9456c9fcf75359ac31cee : SUCCESS in 3m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd067464eb9844948084e78a556ae61e : SUCCESS in 1h 11m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83afe17b56d84bdba95cab98ed55efed : TIMED_OUT in 2h 37m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c778752fb2204c23e290d39bb5db0567328abec4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 18:45:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a501d61b926c2efa1ae96d1c9fbdc1a41c7d7dd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:10:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4896f18e8a40d9841fdba0f160ae7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a775864f796408d8a8cd9c40a12a387 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f855c0998b3f48929f8d1a631b082127 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf1efec3dbcd4c38842389178a5fc0e4 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd26767b35d64527acad49efd5114abd : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eba16f226f7b427db4103edcddf0a800 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/182832e6db4d490d9e669192fbb98ea5 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8e15a3693864334834dc410fdd8de05 : SUCCESS in 3m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3b81d3aad70a40aa91613663a483338e : SUCCESS in 14m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed373e5fb7a6417da98de60f0dbe6558 : SUCCESS in 8m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/856b56f7e39b4e1aaf444ee8a5855742 : SUCCESS in 17m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f1e752d6ecee4341801c69a5e4f7cb1a : SUCCESS in 19m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1a515d6f07904f609e1d64c84c0606cf : SUCCESS in 25m 34s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d08e18a81feb40eeabc11e7c1b1cc03f : SUCCESS in 5m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/65bb42811e224da88f2abaf8a9bdff35 : SUCCESS in 5m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/faba99d141bb4baebd506af8facedcba : SUCCESS in 5m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ec6b0d49d0ac4915b6ba61be7b3a4759 : SUCCESS in 10m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d1929ab7b7114fc9a5ceefb740ef6444 : SUCCESS in 5m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3a999f46027b432c9ba85f4732bd6a4e : SUCCESS in 5m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4c053ed49db546c890f7682b5625dbba : SUCCESS in 7m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c9af75866c1a4b09b6ce9b691c711ddb : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/056a56b5f4064e3386efc57ad4ea5d27 : SUCCESS in 7m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/255a89571f7e4623a6a319498b032649 : SUCCESS in 8m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/443b33999e354137a3ce8ae69ea7391d : SUCCESS in 11m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4581ada62aa24494a0c2def36f49ed4e : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3c9729968ec94715bc56cfb4f7e91f10 : SUCCESS in 12m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7d4fd61eafb74ba3b34099a0807acc12 : SUCCESS in 4m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ad0beae7c4804eb3a36814cdbc65a5a2 : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e2d7838e97ea4fc391c88510c818cb83 : SUCCESS in 11m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/757641018a824baf9f19491839508eec : SUCCESS in 20m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/252c295fdc5d4e0282aa6e2bae10ce81 : SUCCESS in 8m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e6b988eae1374ec19a4e757f6488ac6b : SUCCESS in 11m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/68d8bc0b078f4a2f92ea2d930014dfc3 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ff94b704926481a828adb3f3e5e2201 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbfa5a540d784233892308201658b87e : SUCCESS in 27m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/29a8932f5ff3407eac6572f4e93a096e : SUCCESS in 8m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9f352ef7732744ba831bf806b46542a1 : SUCCESS in 3m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21adf106f018423fac86029251ef0118 : SUCCESS in 5m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a55ab530c91b4575b5c860138180b7ad : SUCCESS in 16m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d08effe6717443af93ee80a620997c66 : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d4e8c6d86534bde8e9d0ed1f3bab8fb : SUCCESS in 1h 04m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1771b4a5fd3a43faaf0982b004562d56 : SUCCESS in 1h 22m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b3bf31df1ad2879726c001d4d8e14ba63f70f5ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:10:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e235e2a37b762bf5a2e5858c95581d24ded70f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 00:16:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36900512eeca47bcb81122cc24351cd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67bbe3f84065441c98ac918eed9dd155 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18b3a47bcab4ec1a29b5a06c74149dc : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0556e73474df4aa88e11a270e4429de7 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/826b645ca13643e8be13667e9d022950 : SUCCESS in 2m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f73b9e34346c4377b78db53f84f258e0 : SUCCESS in 2m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86b1be706c8a462e83401a8762761a72 : SUCCESS in 2m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/43f756a8f2954d7e9af1127c8118d730 : SUCCESS in 15m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cd921b62adb142a8a77d53c93e27a796 : SUCCESS in 4m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6f7dd9bcbed24b15b86df3f39992655e : SUCCESS in 7m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2303f4651c2b4a66aeb915d6fbd0d628 : SUCCESS in 18m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5922371812ff49699e708b7bbaed3e76 : SUCCESS in 8m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/880c97d941bf4aa4b32c0de1bc49fecd : SUCCESS in 5m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/96e52a7a21d343e18f3933856d3a779e : SUCCESS in 4m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8c4f9a1512824e5a95157726f2e2de7d : SUCCESS in 7m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/47e2cd2ce2564cc6897d5faccc36a7be : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8be24ae4f8a044bca9dba96befb3ee05 : SUCCESS in 7m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3a67b4e42848490bbfdbffdf000253dc : SUCCESS in 5m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/169bef8865974034aee146de3d12bee5 : SUCCESS in 10m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b654288f552f42139b8cc68fce7fcdfc : SUCCESS in 1m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7e71e16046d543118245f8f590e5156d : SUCCESS in 11m 13s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/54460e7a536244eeb530368a754a6c05 : SUCCESS in 14m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/077b4b171fd247328a2a966440a3377d : SUCCESS in 7m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5981bb2a6cbf411c8aa4b8b7031535cb : SUCCESS in 13m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a5b124166ecc4773a05a7e2bc6245d43 : SUCCESS in 2m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee8a6549d4ac447c934234167847da15 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/419c0e27aef342b1bc1cd1b1cbb01a28 : SUCCESS in 24m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f7bd550cb3de4f6181a1a05363ae1aa3 : SUCCESS in 3m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c2738bda71cf4f759fa002dec1ff09c0 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d6b12a637b049c5ab8503031d385f8d : SUCCESS in 4m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/306fa4bb583241b0a1be7581bc6c5cca : SUCCESS in 11m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bfe090fe0ff54f90b79c950e84eaa6bb : SUCCESS in 6m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ff730308e364aabb7d7f557ddceb58f : SUCCESS in 48m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1ff2077f6f64ed59ea0df59adfc6418 : SUCCESS in 1h 20m 11s","accounts_in_message":[],"_revision_number":2},{"id":"f6ffb023edb62cd3084f6e372dc0f94e4de70aae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 00:16:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"93703e25e142212666ecdd3462c1090adb5d932c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 00:17:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23e754d4c2704dc69e035a40d8632356\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a4a21c6a1b4bc8947960c69d006ab1 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c088df69c8337dead30631923ffe934d86216682","revisions":{"e022b51730d7a7361ef08a248c9256e40d8c7463":{"kind":"REWORK","_number":1,"created":"2025-07-07 10:42:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/954219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/954219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/954219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/954219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/954219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/954219/1"}}},"commit":{"parents":[{"commit":"1637513a2868445757fb69f9ecad8bd9011a4102","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1637513a2868445757fb69f9ecad8bd9011a4102"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-07 10:32:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-07 10:32:17.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 4.4.0","message":"update constraint for python-blazarclient to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: team: blazar\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I8986120cea37158b602b39974c1a111b3e9f2ebb\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Matt Crees \u003cmattc@stackhpc.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie2b580722611b2c929fe09bff1d97f12749b559d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e022b51730d7a7361ef08a248c9256e40d8c7463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e022b51730d7a7361ef08a248c9256e40d8c7463"}]},"branch":"refs/heads/master"},"c088df69c8337dead30631923ffe934d86216682":{"kind":"REWORK","_number":2,"created":"2025-07-10 12:24:02.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/19/954219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/954219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/954219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/954219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/954219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/954219/2"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-07 10:32:17.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-10 12:23:41.000000000","tz":120},"subject":"update constraint for python-blazarclient to new release 4.4.0","message":"update constraint for python-blazarclient to new release 4.4.0\n\nPin veresion 4.3.0 for Python 3.9\n\nmeta: version: 4.4.0\nmeta: team: blazar\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I8986120cea37158b602b39974c1a111b3e9f2ebb\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Matt Crees \u003cmattc@stackhpc.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie2b580722611b2c929fe09bff1d97f12749b559d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c088df69c8337dead30631923ffe934d86216682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c088df69c8337dead30631923ffe934d86216682"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
