)]}'
{"id":"openstack%2Frequirements~954232","triplet_id":"openstack%2Frequirements~master~Id18275282748d9a34f097b4781af3962613a979f","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-07-10 16:14:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id18275282748d9a34f097b4781af3962613a979f","subject":"update constraint for keystonemiddleware to new release 10.11.0","status":"MERGED","created":"2025-07-07 11:44:48.000000000","updated":"2025-07-10 16:15:07.000000000","submitted":"2025-07-10 16:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954232-new-release","meta_rev_id":"9ff9c83a147e2ca994317f639b60e67362195e3e","_number":954232,"virtual_id_number":954232,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 16:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 14:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 14:02:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 13:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 14:02:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"94ce1efe7eaf466e64113e5bb64012a76a434b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-07 11:44:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ff7e489e4b6285ef7d0e8a3b0acf855b57640b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 13:07:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2224b60797854f75885148dc0029dd4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97a6a30a7a74b1e9ccf8c8d283ccc20 : POST_FAILURE in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1672d80ac64d01bf5e282a73149678 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bd78ac35e2c4fcc8b4c715cb62f5dfb : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c86449479c01403fabd4e6ff9668e2ad : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45c24eed978649dfb44403b4d86fa84f : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5334ba18ff2d4005be36ca30007c03d0 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a352315eed646979cad8d3d453d5cde : SUCCESS in 2m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d00c3980272045baa0018cede1789ec3 : SUCCESS in 10m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/193d8954ddde4d5ca9479b0ed3ced81a : SUCCESS in 9m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/15383f64ea9d4973810bb0523c2f7f2b : SUCCESS in 16m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e1c582b6f1294f1dae8ca419933b4288 : SUCCESS in 17m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1e73373be0ee4caaa11a372128db7fff : SUCCESS in 25m 10s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ac5101e7d99e46a8b347ee85f4c4cd3e : SUCCESS in 5m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5868bdec76564da0ab4e45234b2f4b7b : SUCCESS in 4m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/28184a60e98c402884d57336f540c5fb : SUCCESS in 3m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/03dd2c0e8afb44c59dd2ef68c1e01a3b : SUCCESS in 8m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/212573ff6a8e4ced8ecc7253b641d498 : SUCCESS in 4m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1e2914ce49724955a1eb12ba73c09c23 : SUCCESS in 2m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/45f20334976d4a9a8e6940a66e8c6bc3 : SUCCESS in 7m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/89eda58119ea484e84c58993adb21df1 : SUCCESS in 5m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f3f6941c89a43caa1c446ea42845987 : SUCCESS in 7m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f753f1c9334944b8ad300fa4c5307cd2 : SUCCESS in 7m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9561385eb2a549669a23918337322733 : SUCCESS in 12m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0b48e3229e6347f7b74cec6a66b63d24 : SUCCESS in 1m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/060ae241eb11462a9f5b3b3a324c52dd : SUCCESS in 9m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/319d766bb1a5416f87cd397639fadabf : SUCCESS in 4m 18s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/799f686ea6384b79858c545c83f257fe : SUCCESS in 5m 24s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e2477d0e5e5444c7b3822301ad0775a9 : SUCCESS in 9m 35s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d714e1ac804048389826d445ab574786 : SUCCESS in 15m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d004b19023194a2ba21ea2c8ebc9b002 : SUCCESS in 7m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5a7943c667254b88beac35dcc0265a79 : SUCCESS in 12m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/be5d931b57ba47d7b9356a225f02e535 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8620698863a47b88a6cb5dfe0506da4 : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a0a26ee60ec4973b49fc9d3baccea98 : SUCCESS in 22m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5b490cd15266463482c3fbeb0e494922 : SUCCESS in 5m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ccf188b8a651411486dee0789e26262b : SUCCESS in 4m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73aefe26914a4ca0a73f0233da6f6795 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/efb6d5f53e61434e8be3e1224de1510d : SUCCESS in 15m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cb8e8defa581431d889a0a89da473aa6 : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d65dff0b8efd420a82d5fa5c0da246c6 : SUCCESS in 48m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d962960ca5c74bbba9256b4c2e432833 : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"23ea291621f48d1f075f6c3bd3933d9307285611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-10 12:20:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"55ceed18dc2dc1cb63c7d9406c0c18918fcc0733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:00:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bbe1524ee7e4e19a9c20254dec29425\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbfb2192d384fb79e51b1fc1910d085 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd127a780204de58571959dcadf0ed8 : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a95691866e27402bb8585b2541886942 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2891a18d558400bac8df014920e060e : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2bbdbce1fef4bce8b25e776c84ce373 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfd34f977b5843c79eb6399aa93e3983 : SUCCESS in 1m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c26b5c58b7344aee8bf3fbf296811e28 : SUCCESS in 3m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bdd82311c777454d9f0f0b4eac896bf0 : SUCCESS in 16m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5a587e03f7c44851ac3a1e17f2875d3a : SUCCESS in 10m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/683aa405fb234faba559c6de97470275 : SUCCESS in 18m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/57dced25aaab4e038ef2a80691d833c0 : SUCCESS in 15m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/01e05c46e14a4e9ea87e1f1503f45e71 : SUCCESS in 24m 36s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f57db45f7ffa46ff9a36ac128f6dc128 : SUCCESS in 5m 09s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4b246226f22a46dbb87e35a72b7d8529 : SUCCESS in 3m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/03d57ee9615b4a6f9a94bbfe5df250cd : SUCCESS in 6m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cc2389c2d1ff42e291156ee657d01e4c : SUCCESS in 10m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/71182472fcb04908bf8e50d238278db6 : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/58a449cd86c143328b552e124a440950 : SUCCESS in 5m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cae8422dc56547ca92535e462b8829aa : SUCCESS in 7m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b5367829dcdc4f45867e77ad6e123089 : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0433de7f382492489b52076c44f44f8 : SUCCESS in 6m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/226b809dcb9c4709ab525ef0dba15b87 : SUCCESS in 5m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ccce57a0a1f44ce3b41d75846db7aa64 : SUCCESS in 12m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/71970ba7a790428da1786c78bfc6482f : SUCCESS in 2m 48s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/47cac222d0ed4ab98d366f6e3b1d64fd : SUCCESS in 9m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3e3e3fae25d942d294064619b6536ff5 : SUCCESS in 3m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5aefa8d3fb5348dbb31c554ef70b6431 : SUCCESS in 5m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e926d65b32574187bae6ae03f59e6a2b : SUCCESS in 11m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/97381e4fa6324befb7303a36c403c7e8 : SUCCESS in 13m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/13d2d435440c44438a3d1c3f31f515b3 : SUCCESS in 3m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f5c98135d3714035a60373a196cedece : SUCCESS in 12m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c412e98195ea40e38987160b0504fb69 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01dd6536de0241418d13698efb21baa0 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/91803bdc4edd4a4b902d56a532420a3f : SUCCESS in 25m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/19d394c63ad4490ea96b883edc98bb1b : SUCCESS in 3m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/af3b487596ab418ca9a40bdf5185e562 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b72d45509c34883acd2de2773fc79a8 : SUCCESS in 3m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0d52539116a94e5a8604f52cba21a342 : SUCCESS in 13m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/62e3d82e3be6477dae1ecdc96e1824db : SUCCESS in 3m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d880ab464aa47048e631b6267b57e4b : SUCCESS in 1h 01m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f5ad273a7674bc1aeacdd90b323249f : SUCCESS in 1h 36m 45s","accounts_in_message":[],"_revision_number":2},{"id":"70c486e534a6794e73ce0fb61879dc6c61a7472a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 14:02:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2eda446476d4f25bbb1a1647fe1f52afc39202a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:03:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12441f7875aab2068218ff63c4a0a5ab99f647d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:14:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c4be9fc6a6e42249a524dfb6697d7ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a2b80b27f84124a5f7be62cf683710 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277ef8c2a165409cb4c210d75908c7a7 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9234bc7bf2e14e0b82840475e3c05765 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e8fbf09e6b841a08087e4ad108e24e8 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/637cc98db2e64ab087a0bcbde33292e3 : SUCCESS in 2m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3e81ff6086e4bf0ac10c84ff2a2e6d2 : SUCCESS in 3m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ffd2bc1f42834bfea88bf2d04b53d413 : SUCCESS in 19m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aeeee9021eb94f128fd20eaf6311e9fd : SUCCESS in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/be3da1c957ae41699ebbe182ad5f9d0a : SUCCESS in 21m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ad3eb30488a3401cb9c3966007f3ee9c : SUCCESS in 19m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/50fecf78804e4038a09cabbaae22cdf0 : SUCCESS in 12m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f53815f4ff5b4fa2a1000fcea83fa236 : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/505e9da7d50143049ad469055431c231 : SUCCESS in 6m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cb7626b3297040119e671e33d41c3c1e : SUCCESS in 9m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/513c3209ecc94c339382e03e54880037 : SUCCESS in 13m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41c8531f8f5549af88fa4f80ceb0322e : SUCCESS in 6m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/540700451dbb4d58bb83fde2ec34a6a1 : SUCCESS in 9m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1b56980e47c34708b18e6a32910a58bf : SUCCESS in 13m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a8075f9f917c4f23965cc46e6c4ed644 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9f59dba683de492c87af077c82b0b66b : SUCCESS in 11m 52s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/31b8ce90399b42d9a803f2c727273172 : SUCCESS in 30m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fcf69d8acb044be79d8bfdeb4ea5348e : SUCCESS in 8m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3bbeb46072d643808b27d43e7815869f : SUCCESS in 15m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e04d437cec12491caef8a8ff8f85f6d9 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cbe34ce31942421d83434b9815d3aff9 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae6b31afaac04a538b38a1afa2f3ca0d : SUCCESS in 26m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4ccb2c1af19445e4af00c644dd2226fe : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/70174bd47dba4a9090ae9fc8ec331330 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cd16103a06647ef8359497624814f0b : SUCCESS in 6m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6335c47961264e65b966e0170934b88e : SUCCESS in 16m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ce36551c420d4e418bff324e719bdc8f : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dc5e1c786fc45159c715329edbb991d : SUCCESS in 1h 02m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/605b4fb701d84b6c9ab4fb63c10cc789 : SUCCESS in 1h 51m 29s","accounts_in_message":[],"_revision_number":2},{"id":"73396cfa602f8333d20c46e2783392d07005731a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:14:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9ff9c83a147e2ca994317f639b60e67362195e3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:15:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b28533055abd4b1d834749f1349b30d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973c8014a6ae4c77916f8a6942c5e5da : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"089ae6cf535f356b545de232c4d8f92485af7526","revisions":{"3e2ac041f581b97cd1d578e31186e0a3af77ef4c":{"kind":"REWORK","_number":1,"created":"2025-07-07 11:44:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/954232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/954232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/954232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/954232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/954232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/954232/1"}}},"commit":{"parents":[{"commit":"1637513a2868445757fb69f9ecad8bd9011a4102","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1637513a2868445757fb69f9ecad8bd9011a4102"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-07 11:34:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-07 11:34:47.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 10.11.0","message":"update constraint for keystonemiddleware to new release 10.11.0\n\nmeta: version: 10.11.0\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I087a992c4a1d2782bbcf8ad44bbb4fba118f069b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id18275282748d9a34f097b4781af3962613a979f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e2ac041f581b97cd1d578e31186e0a3af77ef4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e2ac041f581b97cd1d578e31186e0a3af77ef4c"}]},"branch":"refs/heads/master"},"089ae6cf535f356b545de232c4d8f92485af7526":{"kind":"REWORK","_number":2,"created":"2025-07-10 12:20:37.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/32/954232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/954232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/954232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/954232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/954232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/954232/2"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-07 11:34:47.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-10 12:20:20.000000000","tz":120},"subject":"update constraint for keystonemiddleware to new release 10.11.0","message":"update constraint for keystonemiddleware to new release 10.11.0\n\nPin version 10.10.0 for Python 3.9\n\nmeta: version: 10.11.0\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I087a992c4a1d2782bbcf8ad44bbb4fba118f069b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id18275282748d9a34f097b4781af3962613a979f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/089ae6cf535f356b545de232c4d8f92485af7526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/089ae6cf535f356b545de232c4d8f92485af7526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
