)]}'
{"id":"openstack%2Frequirements~954572","triplet_id":"openstack%2Frequirements~master~I5863f40e5f41fca6bd63e492b692dbd1b75f3025","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-07-10 16:40:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5863f40e5f41fca6bd63e492b692dbd1b75f3025","subject":"update constraint for python-glanceclient to new release 4.9.0","status":"MERGED","created":"2025-07-10 09:29:09.000000000","updated":"2025-07-10 16:41:38.000000000","submitted":"2025-07-10 16:40:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954572-new-release","meta_rev_id":"41974f0df543324caac2b4fb0239aecbbaac3dd5","_number":954572,"virtual_id_number":954572,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 16:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 14:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 14:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 11:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 14:15:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e13271eaae514154112301b058bc30c77780539c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-10 09:29:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ca2e59c405da51c8f50749f9d56600459fab6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:07:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0162e2bcb31a42d4814d997cadeb6577\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de026853b6a4c73b9d6dcb22fcf510f : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70793e9691be4f5ebe0fb806ffe0fb0c : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3854e9d8128c4505b8c66166d1b8828c : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/745519e80cec4db9a671e065056a8c31 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f4c2351af174e69bf158e878f09efe7 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ca1dc08172b433fbd53009ffb95eb0c : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba3fa1d010cc4bb196c7cb2cd6ec4b94 : SUCCESS in 1m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ed289c50f24c4148a3b9f335990d9bb8 : FAILURE in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6bb46b278b3a40ef921510e79ceffad5 : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6c67f31148db43d9aa21f2decf1df6c5 : SUCCESS in 21m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7bde41b45c014a63bba4260054ee6f2b : SUCCESS in 14m 40s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5398f47150ed4e3bbee78a99fdefb3cf : SUCCESS in 22m 18s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/859e6b3bebf447d18ac602453aac3d41 : SUCCESS in 4m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1802839f279a4b64a67da4d82a9a0e75 : SUCCESS in 3m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a46d8abaf6f2423f8a2f9a64a717b4dd : SUCCESS in 4m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8be6ef73743e47b78b2eeca18966c2b8 : SUCCESS in 12m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1249ddb4cec540dd8ed010ed8986051c : SUCCESS in 5m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6a02e59f1a24495e82afe6f5d4991b33 : SUCCESS in 6m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aeca52df33d94e51b01fed678fc6dee5 : SUCCESS in 4m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/03404df5d11745e9a62697adc1a284a2 : SUCCESS in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d08cffb2d9e44f4ae39a16dcd31b41e : SUCCESS in 5m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1ea05388ca2844f8b040ea2425fbc11c : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/551515a15a574f1aad7e4d85e986cf71 : FAILURE in 10m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0136a2366b5349c7b9724961932c51db : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7468d7bf838045769f14711f07d49eab : SUCCESS in 10m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4f44532372a74faba9561869cdd5ad59 : SUCCESS in 5m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9ce6a2908b5744fc8c773ba0a91da711 : SUCCESS in 6m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/65e2c6e40b3a49369141cfadbeadced3 : SUCCESS in 13m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a447c8e1c0a8409ab744b5f8420e322c : SUCCESS in 19m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ce44edc836ac400192af7879f06d46ac : SUCCESS in 8m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d7a8adbbe7114682a4ac00e0f0ac8d4c : SUCCESS in 15m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f4e63130bc3043ccac5bd4ccda956864 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/71b5ca81756e4275b5f188b3eca0b81f : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/823d8e62deb84f8dba53109deb1d55b3 : SUCCESS in 27m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/721a1cb31d094a218a44eb43f5527ebd : SUCCESS in 7m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/883a198b714b4c36984af06ec5bf4a0f : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2e4834d7abf4bf29a3fd9630c9af3f8 : SUCCESS in 6m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e4f173766239412394fa04102a598b6a : SUCCESS in 17m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7a828bef6a734454ac8abfeafa297580 : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2c030e15eae426da44fc50ee3b22b4e : SUCCESS in 56m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd9db7a245384a3481dc640ba5413fc3 : SUCCESS in 1h 36m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0eac99cdd1b36b810c77912b5d8fa839d639d49e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-10 12:17:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8ee541f8a7f01c2cfada6993ec89b0cec0a8e5a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:13:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc8491481bb2466fb3c2f1d35e29055f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f787ce28ab9d481bb638a93d0c353774 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a0f8567bd440b68779bd6d62c32247 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c67d39ff787d485fb1a850806a927508 : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f633669b2d7c4f8c83967756338a2106 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bb028d57fa54baebe71997670b0ee18 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5423c00ad4504dec9db11ad62336177c : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/950a774c7e214a12b4e1ae3f1a4b2826 : SUCCESS in 4m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5ed2282c48f54bd280c972ab5687694d : SUCCESS in 6m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f35055459a748a192e968f4436a4124 : SUCCESS in 8m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c2dea076b8974268be86845ee2f540ac : SUCCESS in 18m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0d245abc093248238fb2cd1ea22eb206 : SUCCESS in 21m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/315205380bf64203bb00af17aa2f1cbd : SUCCESS in 14m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0f4cecfc5f344d6d8f6238698b2e54f2 : SUCCESS in 4m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9dff0c1602a24b5882bc5198247002d9 : SUCCESS in 3m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fc100bad3e76402080aaafddc3772b01 : SUCCESS in 4m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a8e8e66325ba42e89778f7c117133502 : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/02c05f48d65c4f0295261d5b754bac27 : SUCCESS in 5m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f80ca49f862c4c56a80804befaf22ea1 : SUCCESS in 2m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5db701de90764185aa758d329c9cd0e2 : SUCCESS in 4m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6fb18e9f70874cc3af491569d92f40fb : SUCCESS in 15m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e880e986e878469790dca77cbb8bbbcb : SUCCESS in 5m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/833b0ea936e4440fbcb3e7713f6f33f8 : SUCCESS in 5m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5646a32b38ae453a8610c784c688cb67 : SUCCESS in 12m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5219486237b547ba9f1b24cec532cb2c : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8cc1af16b4d64c3891e8833d3c0f3414 : SUCCESS in 10m 50s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f325761d69da4169958c3f92334ecf0a : SUCCESS in 5m 00s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/33fcf3341c174a448ef37d3022dabc73 : SUCCESS in 5m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/631aa86717dd4863941347a991435a60 : SUCCESS in 11m 15s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/41734be1a1bd44b0a61f275662afebe4 : SUCCESS in 23m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/64c0c4ed3db24a0eb8a34bbb1ccc2530 : SUCCESS in 3m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/066aa9f0cab045a3ab7783cfa5284314 : SUCCESS in 15m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/86af78a9fb524785b7fef41aaf3af0c4 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57ee3ef69f0d4c50a068c6b807cfbab7 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd955c3edf4a4ecfb7874562e7385ca5 : SUCCESS in 23m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ae338fddab304a419927d96724da1a45 : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5a225fafd9bb413b8df424cc823640cc : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea38030b5678488b89ac6699fae13cb1 : SUCCESS in 7m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f1df701372b64956b230ab4d79a3d967 : SUCCESS in 19m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/094851b0980a4598ab1e15c6a806a50c : SUCCESS in 4m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c366615704404eabb1b04e24b83c0e83 : SUCCESS in 52m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69590614d1364ff48789f3de87c4d754 : SUCCESS in 1h 35m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c9c7f0f79ef12cbd12aa02ac16b36a6b7727768d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 14:15:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e187cee26361177adbb864b0a4baae3315c2ac40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:16:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a8fa1dd4b556dfca2778f5400d06a116e7616a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:40:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fada5ce5e62e48908da4f0d4ed059b82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c2ff651be6d456e855810eefa4a305b : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7db59281b8455b8dfe7590ea3d9d40 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b54d564910a42d790959fbd009dca2f : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a17a7d289a304335af1a98dc27ce9a76 : SUCCESS in 3m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3083d37a66bb4262bea2dac1db42c17c : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4069d358cc6f46a08e83b7c19e7000d9 : SUCCESS in 5m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/73c967c1ee994790a2254ec0aa695a38 : SUCCESS in 17m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/616c4fc5b45345b4ac8eb4b279e36f54 : SUCCESS in 8m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/95757f687af54c33ad1e59a00834e3df : SUCCESS in 18m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/eeae25ed030c449fbc67238178fc0be1 : SUCCESS in 9m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/341c38eaa1bf4f1890f409db1011605a : SUCCESS in 12m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2792dffc49d04560b2366c49ae0f8bbd : SUCCESS in 5m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5619216c24b44935ac332be35c2fd3ba : SUCCESS in 7m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0ff7aaa89f2c473fb82ec2ea80b4292a : SUCCESS in 10m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f9eb8b6b219040279f090dece279061d : SUCCESS in 14m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c643cd2390a044f1a0c4d0f06fd741db : SUCCESS in 6m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/04ca062e1cd0465681e38dcede1a8174 : SUCCESS in 9m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a3c5f6a7ad684860beab9443091b2131 : SUCCESS in 7m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a4cfdd71e5c64816b41fdf049fbe12fb : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/78d9d6b34d7648b78d3d3437274c0aac : SUCCESS in 11m 51s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a0b7a17808eb4b8eaabc92218e264f6f : SUCCESS in 21m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/65113c54661044c1912d2357c59d4198 : SUCCESS in 5m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/50870a9e66944852820c455620e8bf6a : SUCCESS in 15m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/103883a48d9540cfaa895a2b6de290b9 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/267eb70f714a4db48fe8daf1d0bcc0d2 : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87bd01d8ec094c30afefe81affefe5d2 : SUCCESS in 27m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fd59ebe6bcaf4add8f6eb65af536d352 : SUCCESS in 7m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3be8d7c762bb44828edf7aabc211b4f4 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9464940f69ac464fa5d0b980dc55f9f8 : SUCCESS in 4m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0db85cddd8ea4203b5039dcd31a75a61 : SUCCESS in 6m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/345438f09998471ea4ed0b28eea7bd4d : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c2eadcd06454f36b578fb7d455234cf : SUCCESS in 1h 01m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7e281a6a3494300afa9bc4d727bfabf : SUCCESS in 1h 34m 38s","accounts_in_message":[],"_revision_number":2},{"id":"56e9426a3f5e2fca9cd2a26987e533a0ac5afc92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:40:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"41974f0df543324caac2b4fb0239aecbbaac3dd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:41:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/291409512f0a4eb0afd618868596628f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f52450f8cc4d98bc1244c17f48d180 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c38f1c91199046e259375195084082e638c1f299","revisions":{"52cb95f0a05ffac4921e21379838fb6d19646442":{"kind":"REWORK","_number":1,"created":"2025-07-10 09:29:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/954572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/954572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/954572/1"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:19:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:19:07.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 4.9.0","message":"update constraint for python-glanceclient to new release 4.9.0\n\nmeta: version: 4.9.0\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie4e22f4b84214d8fa85ad6221a452580ea67cf95\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I5863f40e5f41fca6bd63e492b692dbd1b75f3025\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cb95f0a05ffac4921e21379838fb6d19646442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cb95f0a05ffac4921e21379838fb6d19646442"}]},"branch":"refs/heads/master"},"c38f1c91199046e259375195084082e638c1f299":{"kind":"REWORK","_number":2,"created":"2025-07-10 12:17:11.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/72/954572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/954572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/954572/2"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:19:07.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-10 12:16:57.000000000","tz":120},"subject":"update constraint for python-glanceclient to new release 4.9.0","message":"update constraint for python-glanceclient to new release 4.9.0\n\nPin version 4.8.0 for Python 3.9\n\nmeta: version: 4.9.0\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie4e22f4b84214d8fa85ad6221a452580ea67cf95\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I5863f40e5f41fca6bd63e492b692dbd1b75f3025\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c38f1c91199046e259375195084082e638c1f299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c38f1c91199046e259375195084082e638c1f299"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
