)]}'
{"id":"openstack%2Frequirements~954574","triplet_id":"openstack%2Frequirements~master~I1c6d5c26af04ffd28d0a5f0fd81a3b301bb1a6cb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1c6d5c26af04ffd28d0a5f0fd81a3b301bb1a6cb","subject":"update constraint for sushy to new release 5.7.0","status":"MERGED","created":"2025-07-10 09:36:56.000000000","updated":"2025-07-10 16:40:04.000000000","submitted":"2025-07-10 16:40:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954574-new-release","meta_rev_id":"d3dcabb115cea77d435791bf2a5e7af15bc99e23","_number":954574,"virtual_id_number":954574,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 16:40:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 14:12:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 14:12:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 11:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 14:12:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"321b680a1c762516fbba16cd3f19ec368e4650a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-10 09:36:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bc1554f0d1d4b0dc95f345fb2ae4a51688d39bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:20:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a89a5c8f045042f7947c31cca1e044ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9252d14e8443ee90281c874203a9ff : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0275d8ab294e6088127f1a20c41c19 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a794dee4304457c838f1eab1cc48d48 : SUCCESS in 1m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36b886cac6d846cd975d58d4168b5e63 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b90b8cb43414b7cbf273035415eec4f : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64f2a24b14254d55b01f8337591da1e9 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9562f8bfbbc145059bf86eb2c7dd5e98 : SUCCESS in 4m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2d2db649edf7461e821dd4da7b9010ff : SUCCESS in 19m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8a1b5c8acc5b4b9c99ca858e2c375dbf : SUCCESS in 12m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/900cf26dc2004ae0bd49e2718004dfff : SUCCESS in 21m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4f9dbb53b009403c936468591b915def : SUCCESS in 23m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/219e66a54f49465888c6b2288c04c672 : SUCCESS in 28m 15s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ccd83c911d4e4a66859f92a44156d7b5 : SUCCESS in 6m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3e573ec275b946d0bc1a8d3624c2d798 : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ea6e3c0718a647a89d23bfc63f6fa8ba : SUCCESS in 3m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/032def65be8241baa568ed4d87fd359f : SUCCESS in 12m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/54ba830fb52943b2aa58160df3e07afa : SUCCESS in 2m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a0b624a5d252481ba0777a6bd5cc96d0 : SUCCESS in 5m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/349be1efe34544d28a0cdfbeac56f291 : SUCCESS in 9m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6641c86713b747aea73452ae62656412 : SUCCESS in 14m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdc2a9c48a73459a8ddb0bd769021ceb : SUCCESS in 6m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0d44731d673d43c7b3fdecfd4e91c959 : SUCCESS in 9m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/eef2ac376a0f4a7fb8556a651b0c637d : SUCCESS in 13m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5339f1fd507144ec950e688c8e2c5b3a : SUCCESS in 1m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5ebb0e7ab14c4276b4806a943d210012 : SUCCESS in 11m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c2e8f7ceec8946239f6fe3645f9e5f26 : SUCCESS in 4m 11s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/88311da1d1d74850805ae7d72aa61bb7 : SUCCESS in 6m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/65b008ab24a44d1ba8966240b01f44ca : SUCCESS in 11m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/78af553115754f4bb8956e337b5d2148 : SUCCESS in 16m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/392707ab8b7f44f4938be8d49a11c1f0 : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fddac0ac51004eb5b6c0bb73507498ee : SUCCESS in 13m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1a7e3e0d1878411786c2119a59c6d4b5 : SUCCESS in 3m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c029d8b788c541dd92aa711177eed20b : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50a0827089744e34a5b732cbde902df4 : SUCCESS in 23m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d95445c0f3c14e728a54681fa75463d4 : SUCCESS in 5m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ca06ef71f15144cc98ca8cbf54ffb2cc : SUCCESS in 4m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8b467d04cf747a99978690a2cbae2c8 : SUCCESS in 6m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ff4b554477424865869621c608afadf2 : SUCCESS in 14m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cc2b213463c64869b322d7cd88351ff4 : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2018d2f9fbcd4eaba33c7dabac5d57f4 : SUCCESS in 1h 06m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3d0ce42e93449fab5c02419366492ca : SUCCESS in 1h 41m 21s","accounts_in_message":[],"_revision_number":1},{"id":"251433c1d74f950e90e746e6d1ccde5000dbaabd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 14:12:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17f375b7b44ffa26f13e3b8b4fb78fdfd9068a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5f33386b9e386ae1484327b6d6784cecf8b9555","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:40:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca36e428c88a4e469f8908e7258b8dc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2e79551841494bb61319742fd15500 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ffe8ee50cb841df9ff8b58aaf0cd49d : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33c722c6dcd24dd4b036342190c8769a : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c91810913e64518a1e786da28369070 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fef546e3b9c6451eb2efff89219b6c26 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ca296d4a63548a8b5fb9c0a1c4e312e : SUCCESS in 3m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae41d86633804523bb3b4d8c8ee2bdc4 : SUCCESS in 15m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/369a1d0f82584d2498267008873f5a46 : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4ede6709b2fe4da0a30136f604809e16 : SUCCESS in 8m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8d877072b90e48d3ab305d6475a724e8 : SUCCESS in 19m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ef4a35f6e6dd4bf0b7b03bdf1b94d40b : SUCCESS in 13m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bda84e9bedca4c93b6781e99c46516a3 : SUCCESS in 7m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ae132176f75e425382030c843944ba0e : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1bceb618fb214763ab0aa97a07e3a031 : SUCCESS in 9m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cfc915ea0bbf4426baf89aad809138ff : SUCCESS in 15m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46ddc1cd66f84d4488b078cf1840781d : SUCCESS in 4m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e8265575fe9e4c06ad1777a11b5ffd0f : SUCCESS in 6m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ee2355f9e68f4461b1c33190ee28895e : SUCCESS in 13m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/64cd938a83c94c6c81f4efcbfc19f910 : SUCCESS in 3m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9967aa2d4cbb4904b1bcaa2ba3f97d01 : SUCCESS in 7m 00s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b50d13641b9d4587a8b6c03636467a92 : SUCCESS in 18m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4252a75c326a4dfea2627f82af939358 : SUCCESS in 5m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/42cc16e088ae4e449999f10ca85c39dd : SUCCESS in 15m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/18b9820fdc98487eb3e15c5fb5c80093 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5d07dc275a24e85854c2cc2ce11e3d8 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64a1e6aa311648e4b74d9c3fe2868b0f : SUCCESS in 28m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/993c47447ebe48acab2553e826c1fbac : SUCCESS in 6m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8375acd4a18b490e94b9fcb6c18b5f36 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5653fab8c4740d49c1b34308e4bd713 : SUCCESS in 6m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c341b73f56084653854b8ddbcfbe1291 : SUCCESS in 14m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4bba45047fe9403eb0f7adeb47370b5c : SUCCESS in 5m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a08d33aa8e4476283a03b7d5802911e : SUCCESS in 55m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97eebc7f4c4741eda8395ea92c01dc7b : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d3dcabb115cea77d435791bf2a5e7af15bc99e23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:40:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9aaf49a3564b870da2b494757483182c3fce7e7","revisions":{"f9aaf49a3564b870da2b494757483182c3fce7e7":{"kind":"REWORK","_number":1,"created":"2025-07-10 09:36:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/954574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/954574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/954574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/954574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/954574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/954574/1"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:26:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:26:53.000000000","tz":0},"subject":"update constraint for sushy to new release 5.7.0","message":"update constraint for sushy to new release 5.7.0\n\nmeta: version: 5.7.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I5da559021645f4752faca595fec6cc9e030c8780\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1c6d5c26af04ffd28d0a5f0fd81a3b301bb1a6cb\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9aaf49a3564b870da2b494757483182c3fce7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9aaf49a3564b870da2b494757483182c3fce7e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
